)]}'
{"id":"openstack%2Fovsdbapp~907871","triplet_id":"openstack%2Fovsdbapp~unmaintained%2Fyoga~Ie6fb95c7411e72089ff13cdba28666f2e1d06efc","project":"openstack/ovsdbapp","branch":"unmaintained/yoga","topic":"create-yoga","hashtags":[],"change_id":"Ie6fb95c7411e72089ff13cdba28666f2e1d06efc","subject":"Update .gitreview for unmaintained/yoga","status":"MERGED","created":"2024-02-05 16:47:11.000000000","updated":"2025-02-16 14:37:50.000000000","submitted":"2025-02-16 14:34:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907871-create-yoga","meta_rev_id":"6ebdca57893194882415ca5debe3ad18e5a71a7a","_number":907871,"virtual_id_number":907871,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-16 14:34:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-02-16 12:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-02-16 12:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 17:03:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-09 19:44:51.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-04-03 16:40:58.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2025-02-16 12:31:59.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"d12783b38bb99e4397d2c5e60836d3a1275db23c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-02-05 16:47:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2662cd4131c2682662cdd4a2ceac4d3919373986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 17:03:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49337fcb61c644a1a0ea0abdb53e726f\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea079d1b0dcb4757b9767a04f6f71b1e : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd8d720b3484533937a62e6bc20018a : FAILURE in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e89966ec3c747f2ab6b548b21534396 : FAILURE in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05f1a293a41c4bbb80a3302044b704ee : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5777f0d24e0841e88d7cd7b0acaf58d2 : FAILURE in 4m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f5a035b7f97c4bebb19f3031b0dbb8a2 : FAILURE in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7bc317fc05184239cef5ebbed0826476843d82c1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-09 19:44:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb3cb7d6421b1df83d85fa37eaf7647293565943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 20:05:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abd840d2efa340e89e2b5e394ec74b57\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2d4ebbf3aea45eda2b40f30acfd8078 : FAILURE in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea46fd6872144faca160feb3f0334f26 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/525e303f0fc24f3fbf70a882fe553179 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ce63a63b1bc4782aaeda043b3ca59be : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea3080507c541088ba31bd15dd51614 : SUCCESS in 4m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1ceb22a2b33a475d8d4ef7feb568ee09 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7bbb225e8fd91b114b52b586ff2fdbc59e6eb614","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-04-03 16:40:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1cb3ee9ab9ac82defb51cfdf1074f9e9d6e19e23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 19:08:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e22de026b03849a38b04cf0ddd0eaea9\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19ede457981c41ab97a60f61ce949325 : FAILURE in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f48c902622148aa866931b94404e1bc : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bf2f8558c1e490ab3a6d7ce3aff1533 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d9fcba32805412b8484f710bddbb983 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215387596876403ab4f09ff707da1921 : SUCCESS in 4m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/860275b3abd6478887249cc88f496c6c : SUCCESS in 5m 02s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/74f10bb827dd4a3885b459ffc9db0ce1 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":1},{"id":"664c4bcdce99973177309f34a4eee98a91da887c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-16 08:28:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53485216b2945c7035de50d3b919d5942d3a3134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 09:14:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30a16d39e3a04e84a89260bda94ca0e8\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0314ad4143704cac9efd813cf59af2b7 : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e028c93cd9bf42dca1c807aac164ece1 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db6a3157fce6477ba496b97f86249964 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97d871ebb9c24f4aa9b71f80c0cab0c9 : SUCCESS in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f29d69a7d3e4c06b41b80f79648d1ab : FAILURE in 4m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/39e2023ffb884c11834986295d784229 : FAILURE in 3m 23s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/882f7f31001a458fae92364c6d59eff8 : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5d38077b685d20d305566034b247de208fa3f8e1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-16 12:31:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"336e7b1f054be30d4649925a110a331c974915e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 13:51:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adb067fa68c24a0699021378d714d624\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/474f5017384041c0a447b93970848418 : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dda1443582e34c7bb67e3a3978f1118c : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce0cfe540e144197a4f7d5efcf474a1b : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2e5f2b45f543518661f5ae1a20676b : SUCCESS in 4m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/14fa6fb47bb7490ca61389281b74f19f : SUCCESS in 4m 50s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c7ddea705ee947d28daf259b08c6f969 : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"27e8a89943837753913c72e4e8bcdf71ffbf17e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 13:51:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"392b92516bb876de38d9aff9cf867502590260c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 14:34:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/869f8b1ce6834da6bf0c4849d280b1b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c079728dd534bb79c9ca7be01a11465 : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a8f7327cf094a4fb70176af6d81544b : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee30f20f336d4b68abc2749391fe8417 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0eb499338e64efa8af9a99163234c8b : SUCCESS in 3m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/601ed1a03761482ca7c582122468ea54 : SUCCESS in 4m 35s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fb0a8821df064b5c8133e23cb44d8c39 : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"610d12fdc4dac02d67c65e134ee77ba45620afc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 14:34:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6ebdca57893194882415ca5debe3ad18e5a71a7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 14:37:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa7c71a48b2b48cba56896c395792e2a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69703e42a7894220b177836fa9ed91e0 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fa7d945ed4166ae2fb326379aa6779bf5dfced0","revisions":{"2fa7d945ed4166ae2fb326379aa6779bf5dfced0":{"kind":"REWORK","_number":1,"created":"2024-02-05 16:47:11.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/71/907871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/907871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/907871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/907871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/907871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/907871/1"}}},"commit":{"parents":[{"commit":"ab3e0cb0d0865417efbf103f44954573a5ba92ac","subject":"Don\u0027t force_reconnect() on unhandled Idl exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ab3e0cb0d0865417efbf103f44954573a5ba92ac"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-02-05 16:47:11.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-02-05 16:47:11.000000000","tz":0},"subject":"Update .gitreview for unmaintained/yoga","message":"Update .gitreview for unmaintained/yoga\n\nChange-Id: Ie6fb95c7411e72089ff13cdba28666f2e1d06efc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2fa7d945ed4166ae2fb326379aa6779bf5dfced0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2fa7d945ed4166ae2fb326379aa6779bf5dfced0"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
