)]}'
{"id":"openstack%2Fovsdbapp~920404","triplet_id":"openstack%2Fovsdbapp~master~Ic6256b4bab48298fe599b7904e9edb0e35b93ce4","project":"openstack/ovsdbapp","branch":"master","topic":"use-log-warning","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2024-05-28 08:03:25.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-05-28 09:20:16.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2024-05-28 09:20:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6256b4bab48298fe599b7904e9edb0e35b93ce4","subject":"Use log.warning() from the logging library","status":"MERGED","created":"2024-05-24 16:22:17.000000000","updated":"2024-05-28 09:21:21.000000000","submitted":"2024-05-28 09:20:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920404-use-log-warning","meta_rev_id":"71d1a6dc0110a9333b3b22e99b274964c79287cc","_number":920404,"virtual_id_number":920404,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 09:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2024-05-28 07:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-05-28 08:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2024-05-27 09:12:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-05-28 08:03:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-24 17:25:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-27 09:12:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-05-28 07:04:58.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-05-28 07:04:58.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2024-05-28 07:04:58.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"ce5866714a6e797f02dbb4921611f0b8dcd3cf9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-24 16:22:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2aec7841b5cd534ba7700ccdea60c3b45449576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 17:25:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f92ef390cb2146929b9d554f39af069a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c82f4353116348599f5560363ebe4cb9 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/314b011b8c094bf9ac3bc0d7506711e6 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98b5871ab4454107b17b28666c0dffee : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af9f47618a254eb5a0cc405fbcac04ee : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c408b5e77cbc47faa497fa7709192260 : SUCCESS in 4m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/55ad18b63af24cd396c41441abb03b3f : SUCCESS in 4m 20s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6d0e5d7dbab54eeeb4a30a1868974f58 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"dfa5f3f9e6217e5573ed827c46c45f371683f64a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-27 09:12:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df03259e66a17ae1b6a0d063da7871b236d1bd3e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-28 07:04:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4a00fbde597b782646cf3936ca4b762e3647fdd","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-05-28 08:03:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"adbda66b5b07ae27d813f201ca83507062c7e92f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 08:03:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f17cc490677b0c692204ec8559b24000a91c0727","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:20:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/963e6f3ee73c4bebaf5431e904abaead\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1bb13dba464f50b65e811d518408cf : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/693f01cb68f645acb4d2d0b126bc9c07 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b241528234c4452a868a425af737fde7 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6507bf1527c74a0e8af98abc8b1ec855 : SUCCESS in 4m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a4331d086acc45da885e36a801da2f32 : SUCCESS in 5m 03s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a606201fc76d4a7bbcade71a10300096 : SUCCESS in 1h 14m 53s","accounts_in_message":[],"_revision_number":1},{"id":"30533b4f42ebfc0eee6eaa706974f95dcaf7c3fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:20:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"71d1a6dc0110a9333b3b22e99b274964c79287cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:21:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c13bd21c42246e8ab0037a28ed0e094\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f81eb49cae64aaabdd31f048e1a25e5 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb8a07aacf2eeff9c91397d208a9c964cec8b4e6","revisions":{"eb8a07aacf2eeff9c91397d208a9c964cec8b4e6":{"kind":"REWORK","_number":1,"created":"2024-05-24 16:22:17.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/04/920404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/04/920404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/04/920404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/04/920404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/04/920404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/04/920404/1"}}},"commit":{"parents":[{"commit":"bccbf70dfe2f8e02c3b5aa82ba1557a35c89f62f","subject":"Merge \"Improve compatibility with netaddr 1.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bccbf70dfe2f8e02c3b5aa82ba1557a35c89f62f"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-05-24 15:54:13.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-05-24 15:54:13.000000000","tz":-240},"subject":"Use log.warning() from the logging library","message":"Use log.warning() from the logging library\n\nEvery time I run a large chunk of neutron unit tests, for\nexample:\n\n$ tox -e py310 -- neutron.tests.unit.plugins.ml2.drivers\n\nI see this warning (it doesn\u0027t cause a failure):\n\nCaptured stderr:\n~~~~~~~~~~~~~~~~\n    .../neutron/.tox/py310/lib/python3.10/site-packages/ovs/db/idl.py:1484:\n  DeprecationWarning: The \u0027warn\u0027 method is deprecated, use \u0027warning\u0027 instead\n    vlog.warn(\"error parsing column %s in table %s: %s\"\n\nLooking at the OVS python library code it seems perfectly fine.\n\nBacktracking, I was able to find the offending code that\nactually lived in ovsdbapp. It is a little bit obfuscated\nas it was re-mapping the OVS library vlog.Vlog.warn() call\nto log.warn(), but I was able to test my theory by modifying\ncode in my local neutron .tox directory and the warning\nwent away.\n\nI was never able to narrow things down to a single test, or\nclass, or file, even running with \u0027--concurrency 1\u0027, so I\ndo not know exactly how this code is getting exercised,\nthere could be a test that is not cleaning-up properly,\nmaybe some day we will figure that out.\n\nTrivialFix but not TrivialToFind\n\nChange-Id: Ic6256b4bab48298fe599b7904e9edb0e35b93ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eb8a07aacf2eeff9c91397d208a9c964cec8b4e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eb8a07aacf2eeff9c91397d208a9c964cec8b4e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
