)]}'
{"id":"openstack%2Fovsdbapp~936877","triplet_id":"openstack%2Fovsdbapp~master~I69933f5536e6cb3604f89802909603a0413d4d15","project":"openstack/ovsdbapp","branch":"master","topic":"bfd","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2025-01-10 09:57:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69933f5536e6cb3604f89802909603a0413d4d15","subject":"Add bfd parameter for lr_route_add","status":"MERGED","created":"2024-12-02 19:40:50.000000000","updated":"2025-01-10 09:58:18.000000000","submitted":"2025-01-10 09:57:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936877-bfd","meta_rev_id":"e9ca0378a03734fe4664672a5b77ada167bebb85","_number":936877,"virtual_id_number":936877,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 09:57:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-01-10 08:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-03 14:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-01-10 08:49:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 20:47:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-03 14:21:48.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2025-01-10 08:49:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b579a2a5a37aab4a6735652531849642f4dda771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2024-12-02 19:40:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c62af59575df39124eaec09598a7b72680ac7afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 20:47:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e2ba1fcca014df1b849e6fdd390794e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd77eb85f3ab44c2be778772f048c236 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dd8cdb604b04d5a9bfccbf4e48bd615 : SUCCESS in 2m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bdbea1f2c1b843d8b659ca7118ee4846 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6336239a33b14465a730a884979a057c : SUCCESS in 4m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/60f1969c1b6749b4b313f9d749b93ea4 : SUCCESS in 4m 08s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fc44c7c46eec4e38a9a17ba726ee1780 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d855cb0e264997ab78bc6515b9a0af8db13bf3c2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2024-12-03 14:11:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97e88f5c5144c26c1ce0ba33d36325a0c7c548d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2024-12-03 14:12:25.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cc1433fb213dafe8ecbdf1866c84689dbc0d9fff","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-12-03 14:21:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3eae2c32e50cbb1eb1bb3579258995d720690180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 15:40:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8633b6e5522e403c8f2ec106b8b060d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16360edb0a1743ecbc258342cd0d628d : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7611d3255063472f93e11a942ab24f31 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4ec6dab84ec452ca40f65d40054cf8f : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1633233255fa4f0d961dbde7d04f513e : SUCCESS in 4m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9a4aee93261f4faab7003059d18db699 : SUCCESS in 5m 53s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3b528ba6b9b94c6bad814b16b3bc89c4 : SUCCESS in 1h 22m 20s","accounts_in_message":[],"_revision_number":2},{"id":"fc7360896b62ff1add26c6db36b74ec45a9ac40c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2024-12-06 17:56:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d433f203ec0a32bb862ab5c0d5ce7177d80e8527","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-12-06 20:28:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6e59290dffec4b3eac6a8d0cf8f36e811de5fad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-01-10 08:49:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b6624e44c4148ce142ca6a0c6af7e49b4695af83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 08:50:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a8e932adc2f4fd1d72a5e1bcf2332eed1324d16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 09:57:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88f5dce3bc994d19bdfe139b172f097a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f26f7bf7c745f6ad28593ea0d6d181 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a5f8379db0841b2b7736d7f718452d8 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/329a89548cf2410cb8a8c25264c8702d : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de65adca667444f9b223ed77c50dee7 : SUCCESS in 4m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/619272df536c4e5f8b66187017e45c99 : SUCCESS in 4m 51s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a22888469494400a9630b0a88a990d8e : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":2},{"id":"854b3a5103f702f09387361dd5d889686ca60ebe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 09:57:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e9ca0378a03734fe4664672a5b77ada167bebb85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 09:58:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0820406e1f3f47d39217fb4bc2bf0c12\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fae0273f8ca493fa12f30317f327473 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7a0dfa26455f37d1e9ef50fe4585954c76db9547","revisions":{"2c5e507d5afb0ddd24809383349ba745b4b7e342":{"kind":"REWORK","_number":1,"created":"2024-12-02 19:40:50.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/77/936877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/77/936877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/77/936877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/77/936877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/77/936877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/77/936877/1"}}},"commit":{"parents":[{"commit":"8b7ce6dbe5baee273159d648304250d0097d3703","subject":"Merge \"[NB] Add \"if_exists\" flag to ACL and Port_Group deletion commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8b7ce6dbe5baee273159d648304250d0097d3703"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2024-12-02 19:25:59.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2024-11-28 23:37:10.000000000","tz":0},"subject":"Add bfd parameter for lr_route_add","message":"Add bfd parameter for lr_route_add\n\nThe parameter refers to the BFD table entry.\n\nChange-Id: I69933f5536e6cb3604f89802909603a0413d4d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2c5e507d5afb0ddd24809383349ba745b4b7e342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2c5e507d5afb0ddd24809383349ba745b4b7e342"}]},"branch":"refs/heads/master"},"7a0dfa26455f37d1e9ef50fe4585954c76db9547":{"kind":"REWORK","_number":2,"created":"2024-12-03 14:12:25.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/77/936877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/77/936877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/77/936877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/77/936877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/77/936877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/77/936877/2"}}},"commit":{"parents":[{"commit":"8b7ce6dbe5baee273159d648304250d0097d3703","subject":"Merge \"[NB] Add \"if_exists\" flag to ACL and Port_Group deletion commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8b7ce6dbe5baee273159d648304250d0097d3703"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2024-12-02 19:25:59.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2024-12-03 14:12:24.000000000","tz":0},"subject":"Add bfd parameter for lr_route_add","message":"Add bfd parameter for lr_route_add\n\nThe parameter refers to the BFD table entry.\n\nChange-Id: I69933f5536e6cb3604f89802909603a0413d4d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7a0dfa26455f37d1e9ef50fe4585954c76db9547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7a0dfa26455f37d1e9ef50fe4585954c76db9547"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
