)]}'
{"id":"openstack%2Fovsdbapp~937445","triplet_id":"openstack%2Fovsdbapp~master~Id20b93d74508a049c75959e3c69a902739a7b032","project":"openstack/ovsdbapp","branch":"master","attention_set":{},"removed_from_attention_set":{"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2025-01-29 16:42:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id20b93d74508a049c75959e3c69a902739a7b032","subject":"[DNM] Adding some logs to debug txn commit issue","status":"ABANDONED","created":"2024-12-10 09:42:24.000000000","updated":"2025-01-29 16:42:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ecc765f2ef1c6247f53c64c3e1dda532c9d0c943","_number":937445,"virtual_id_number":937445,"owner":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-10 12:57:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-10 10:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1931d4996dd7e2146a88b77f0331f32c96665ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-12-10 09:42:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"866c4b6a6dcacd30b33903023bcc407a6a7f80f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:36:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d91c0d371d1d465682c00f686f6b9eb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f9b08177e746599c09944a0f8f2956 : FAILURE in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a54a378eca034aeb8a29b224a53a1e5e : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51da34aa756145efa15a9875b0133313 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccfd454850db4a2abae5238dc82cafe0 : SUCCESS in 4m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/28c64985a4854e638f3d7ea2a4b1713f : SUCCESS in 5m 26s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c92e7bd38289448ca55ce9ee83713176 : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fc8f66fc37344235ef73f26798946412850c4a43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-12-10 11:39:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a79a40c495a9054109b2ffdbb43cd1929e1a2a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 12:57:29.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1eaf63e573e4d0dbd47b28a3097ea8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0405b09807b1487e8131d3a86377058b : FAILURE in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecd88f3c817e4d6dac28ee0ca338934b : SUCCESS in 2m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e509d07d150146c28c165f20e31dc6c2 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c9de6472b0d4b9b90874aeb3979f135 : SUCCESS in 5m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ebddcd2964f840b08f7185ee988aa484 : SUCCESS in 6m 55s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/badd00e130594743b7a9ee24c3bc146d : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ecc765f2ef1c6247f53c64c3e1dda532c9d0c943","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-01-29 16:42:11.000000000","message":"Abandoned\n\nLooks the issue is solved!","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"551b93ca775d45d2377f2a97633180b57a2b032c","revisions":{"ad38c8467f0f5a54caa45a5def659bf1069da36a":{"kind":"REWORK","_number":1,"created":"2024-12-10 09:42:24.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/45/937445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/45/937445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/937445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/937445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/937445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/45/937445/1"}}},"commit":{"parents":[{"commit":"8b7ce6dbe5baee273159d648304250d0097d3703","subject":"Merge \"[NB] Add \"if_exists\" flag to ACL and Port_Group deletion commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8b7ce6dbe5baee273159d648304250d0097d3703"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2024-12-10 09:41:14.000000000","tz":60},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2024-12-10 09:42:22.000000000","tz":60},"subject":"[DNM] Adding some logs to debug txn commit issue","message":"[DNM] Adding some logs to debug txn commit issue\n\nadding some logs to debug another issue with stuck on txn commit\n\nChange-Id: Id20b93d74508a049c75959e3c69a902739a7b032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ad38c8467f0f5a54caa45a5def659bf1069da36a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ad38c8467f0f5a54caa45a5def659bf1069da36a"}]},"branch":"refs/heads/master"},"551b93ca775d45d2377f2a97633180b57a2b032c":{"kind":"REWORK","_number":2,"created":"2024-12-10 11:39:21.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/45/937445/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/45/937445/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/937445/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/937445/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/45/937445/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/45/937445/2"}}},"commit":{"parents":[{"commit":"8b7ce6dbe5baee273159d648304250d0097d3703","subject":"Merge \"[NB] Add \"if_exists\" flag to ACL and Port_Group deletion commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8b7ce6dbe5baee273159d648304250d0097d3703"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2024-12-10 09:41:14.000000000","tz":60},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2024-12-10 11:39:12.000000000","tz":60},"subject":"[DNM] Adding some logs to debug txn commit issue","message":"[DNM] Adding some logs to debug txn commit issue\n\nadding some logs to debug another issue with stuck on txn commit\n\nChange-Id: Id20b93d74508a049c75959e3c69a902739a7b032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/551b93ca775d45d2377f2a97633180b57a2b032c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/551b93ca775d45d2377f2a97633180b57a2b032c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
