)]}'
{"id":"openstack%2Fovsdbapp~941839","triplet_id":"openstack%2Fovsdbapp~unmaintained%2Fwallaby~I010e145f4c5079a80b72c5229dfacc4d6896373e","project":"openstack/ovsdbapp","branch":"unmaintained/wallaby","topic":"use-ovs-main","hashtags":[],"change_id":"I010e145f4c5079a80b72c5229dfacc4d6896373e","subject":"[CI] Fix multiple gate issue","status":"MERGED","created":"2025-02-16 14:01:04.000000000","updated":"2025-02-16 19:07:50.000000000","submitted":"2025-02-16 19:06:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941839-use-ovs-main","meta_rev_id":"92d16f78c853e31df8d63fe139b4237e42fa95c3","_number":941839,"virtual_id_number":941839,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-16 19:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-02-16 17:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-02-16 17:43:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-16 15:07:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69716dc1fbfb3f408a8c206cfc2b105bdb722090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-16 14:01:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08ec56928f0c9a6c2d67e46dd3e929bc65199ec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 15:07:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a73c29a74d9418087cf9557a4496467\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a2a0b4fdb44b26aae5b309988ffd84 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/218c759df92d49d58445ba87eddcd62a : SUCCESS in 2m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b148b38a96d1402faceee30c4d04f77d : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0a9f82d380743258656ad353cc5f3f9 : SUCCESS in 2m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bcba91d9b554160915c707be5b3c9ea : SUCCESS in 3m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/864090fcd11046c68e66dff995314ee9 : SUCCESS in 5m 45s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/db56871c98e042519be099ce0c881f4d : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fccc95712c1a7b72950096fc9c984dce2763bc17","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-16 17:43:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32efa4579e413c09033856182332d754686d0241","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 17:43:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dd5fe3da21e7eefac264bf6e5bb947498311e98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 19:06:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a228144f425240bf959f7cc11738c698\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2b605caa1a4f3ca97049517618df99 : SUCCESS in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8465c932364e40dd9998c4b62fa57c1f : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba814c77e7fc47afbba5fef606481173 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df751fda17f43d6ab2193eb90814558 : SUCCESS in 3m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/228e23061fdb4031bbdf7cf9bb80bf76 : SUCCESS in 4m 25s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4cc14c888a5940fb857a951a49dfccd6 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8f11831563dcfb6e2e3cad242a9740deaf502392","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 19:06:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"92d16f78c853e31df8d63fe139b4237e42fa95c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 19:07:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/350443bd591b4a32914dc64c563c1386\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec7ce31175b4476a7b267cef03908d6 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c49ea2825bd31453cddef300d31e8daf0616803d","revisions":{"c49ea2825bd31453cddef300d31e8daf0616803d":{"kind":"REWORK","_number":1,"created":"2025-02-16 14:01:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/39/941839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/39/941839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/39/941839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/39/941839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/39/941839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/39/941839/1"}}},"commit":{"parents":[{"commit":"65d02f06d2318c20ee01ba760bbb74f6eafabebf","subject":"Don\u0027t force_reconnect() on unhandled Idl exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/65d02f06d2318c20ee01ba760bbb74f6eafabebf"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-16 08:22:43.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-16 14:01:21.000000000","tz":60},"subject":"[CI] Fix multiple gate issue","message":"[CI] Fix multiple gate issue\n\n1. Use OVS main branch\n\ngit checkout of OVS master branch causes an error [1].\nOVS deleted its master branch and replaced it with main branch.\n\n2. Cap setuptools \u003c71.0.0\n\npy38 jobs (on ubuntu-focal) started to fail (see the relevant github\nissue [1]) due to recent virtualenv release (20.26.4; which bundles\nsetuptools).\n\nsetuptools is bundled in virtualenv, so it has to be capped via the\nvirtualenv package. tox also needed to be capped (\u003c4) as gate uses\ntox 3.28.0, but with capping virtualenv we pull in latest tox as well,\nwhich would cause other errors.\n\n3. Remove lower-constraints job\n\nlower-constraints jobs are not supported anymore and are removed from\nmost of the branches, so let\u0027s remove here as well to unblock the gate.\n\n[1] error: pathspec \u0027master\u0027 did not match any file(s) known to git\n[2] https://github.com/pypa/setuptools/issues/4483\n\nConflicts:\n  zuul.d/project.yaml\n\nNOTE(elod.illes): conflict was due to branch specific job template\n\nChange-Id: I010e145f4c5079a80b72c5229dfacc4d6896373e\n(cherry picked from commit f2d79499e6a478bed4f1fd4c412aaa59ff375c3c)\n(cherry picked from commit 1a49de78881ca344763e5e356108b51a8fb0e215)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c49ea2825bd31453cddef300d31e8daf0616803d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c49ea2825bd31453cddef300d31e8daf0616803d"}]},"branch":"refs/heads/unmaintained/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
