)]}'
{"id":"openstack%2Fovsdbapp~952871","triplet_id":"openstack%2Fovsdbapp~master~I96ead59a0b278ab87a2548f7bff1252e1071dfbb","project":"openstack/ovsdbapp","branch":"master","topic":"improve_stringification","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2025-11-13 14:17:16.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2025-06-21 15:07:57.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"I96ead59a0b278ab87a2548f7bff1252e1071dfbb","subject":"Improve Row stringification","status":"MERGED","created":"2025-06-18 22:04:39.000000000","updated":"2025-11-13 14:20:17.000000000","submitted":"2025-11-13 14:17:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952871-improve_stringification","meta_rev_id":"c5a64c45b33170c3f54e16dcba3e9c5f1d516a45","_number":952871,"virtual_id_number":952871,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-13 14:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-11-13 11:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2025-11-13 12:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2025-11-13 12:55:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-18 22:15:49.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2025-06-18 23:23:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-19 05:53:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-11-13 11:26:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-11-13 12:55:39.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"bff492f5d6d6e42b4d7e3b9ef61f1c69030d2b17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-06-18 22:04:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d42850e08244dd808db710a1c63259d0ddea5e15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-06-18 22:08:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"607f5d88502002733379d36a6ef65f7dbfefe36d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-06-18 22:15:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f9ea5409a71073e83eab459a646425bed25a2a0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-06-18 22:18:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d197aa68d3f36fbd74dedeca2d951b9f9b895615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 23:23:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c800f955674a9d850d8f1b7a0bccba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde8e91120e74a58b5d23d423420e527 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca739d815d8b41e49b8513ef9bc27e01 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e0b4e5023b34430baf3792a8324e740 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c4cceafa03e40f6b1e76b70b6863fac : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a859a0795a994011b09d7449f151141d : SUCCESS in 6m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2d582f3b597f4a1093875efe6b282f04 : SUCCESS in 6m 55s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/33801549560b45cdb5507c70fd82c16b : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8ede485b00eb70375a51161574aaa0239d1bbff2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-19 05:53:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e0e5e6ba78cd5169ca91b995f65c04ce027b2002","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-06-19 19:18:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aea457fcd0318f563508222dc65a01ad3a03cabc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-06-19 22:24:36.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"116f5b54a8e34f66febc5f5200657e61194f2616","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-06-19 22:37:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c31ee56762c38c459f3a4c9a22f2d5503808d24b","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-20 09:49:00.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_16688\u003e\n","accounts_in_message":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"_revision_number":2},{"id":"8271b5736ea90b363926e8f1b38a1959de22c0bc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-06-20 13:27:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b996869f6ed0d5c9d45ba21919d3732bef74bea8","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-06-20 13:42:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b8d2a798d402f8448a4c1c37783607ae77026ad","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-06-21 15:07:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08b1022dc71fd8284623333788419ce3313398f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-10-02 02:48:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"38654d17861c329dffa02ac598ee99dad8c74f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 04:07:00.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/681c5994fcc24c1584cadf11d5371163\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b35fff668e494a57887bb9b67040ab6b : FAILURE in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0418187eefc4b278e181270e36eb1c7 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ff2f8a28cb9491486533ef1b5017241 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0661b6cc056d48c8bc9e39d85c7f4461 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a194ff42ff432995fc15fd5ca47730 : SUCCESS in 7m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/157ae126e834402daedc1a4471cc1ca9 : FAILURE in 5m 24s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f37bc5f4d74a4cceb492efa15fcb1a28 : SUCCESS in 1h 17m 46s","accounts_in_message":[],"_revision_number":3},{"id":"17a4023046f271ed6afa3676cddccfe35e4e7360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-10-02 05:04:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a6a54e6fbf5e062cda06cd82158a4677b781739f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 05:54:24.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bba32fa5b1774f5aa1dffa84bf7a46ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa021eb7bdc34edbab8ca88393d160e5 : FAILURE in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f64e987d6d674ade99ce89c4fc097c24 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ebce4a3450c144a38a40c7a91984b69d : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79817815b579490aa77accfdd2213adc : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b3e911951540ef807e27d6a2f14b45 : SUCCESS in 6m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fe0e8475988a40d8af5f434ba4fcf1f6 : FAILURE in 6m 22s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fe38e99d1fcd4265807e542f3e64b234 : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":4},{"id":"971148c34e82a4efb8639ef9ec73995bdfa8e491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-10-02 15:12:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"be8b42ecdceb93036378363e4cb8a3ab1eeb2e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 16:38:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/110eef3e1f404d08a3d7e21b84758eae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4102822f9c0a44128c26a36781833def : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec40c0de13034da38dae9fed3c134f6b : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85d5d42c2e804781bc2ec453c3714188 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/372d79eaa56e4ee1b5c5ee6451a1f79c : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87427ad9073b422887ea5b85c95f8760 : SUCCESS in 7m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e3ea60ee5ce4414e868d00962ddbce0f : SUCCESS in 7m 10s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4ff60260d76446bd97d9c2f078a2fb1e : SUCCESS in 1h 16m 14s","accounts_in_message":[],"_revision_number":5},{"id":"aaa16b4d1103e3c8d1aa38a24789e2f5fce34c43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-10-23 20:51:37.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d725c6e264604d3475a79af64772c27d65d0121e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 22:15:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14ec7d992e634eefbbd917655ba9e57b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b050bec5534b7a8a2b8a3c84ba019c : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6488a961db894a7dac0dc3052fdf43f4 : SUCCESS in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c30761667d854036b18a2ef031aad035 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/573d6278612b4efb849ead432a5150a3 : SUCCESS in 5m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3b87366eaf50473a9880f999441a40fd : SUCCESS in 5m 37s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8ac9100881d64f4e8630c615f8d6b133 : SUCCESS in 1h 18m 38s","accounts_in_message":[],"_revision_number":6},{"id":"6ecb778304cebf4d24f07d2c6b49774601377e66","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-11 14:28:57.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b10abf6e93f2cd50d1d251ca16d5802ea39febc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"real_author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-13 09:54:16.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_5756\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"_revision_number":7},{"id":"f72e2de45cb0237144c2d2d9399a7e54ac077562","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-13 11:26:46.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"257018a2897719b085c48706fd10d2b3eb5be14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 11:33:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fe19b96ce414a65b5b7f880aceb6a96\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b92ac413c6f4f55af61800a5358874a : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/286854f8288f48bcbe2538f5fc2e6c9a : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/953b266c403b48c9800a3b6c6244890d : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7ff7e0043c4eb6a1a68e880c58fc5f : SUCCESS in 6m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/eb563c77df2942d8aace15478b911815 : SUCCESS in 6m 44s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f597c840403240af908ba534b4892aa3 : SUCCESS in 1h 28m 03s","accounts_in_message":[],"_revision_number":7},{"id":"0b9fab6d8baf95b33501f72ca15eb9d59de94076","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-13 12:55:39.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"42afd0e0d4051a252410aa427eae20fff9a1aec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 13:01:00.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"93f5a90cc6f103a98586df7357b718b12304f923","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:17:16.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b5dc31042bc43f49541aecd9bf6b407\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979df984752442259a69de2bdc81fc0f : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ce054d8f8c54b2e82fd00187b708388 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6362a2e5c1164d96bad84a36e8fb1f28 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20800ec15d842b0b15bfd64bf57952b : SUCCESS in 3m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a5c6aa13005a4b1eb7f12eb6058fd620 : SUCCESS in 3m 55s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8e3c55c55bd740bfabf3ab75549a3f14 : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":7},{"id":"23b1fdec77f1ea8dbb088c5c52b052fb3abfe0d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:17:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"c5a64c45b33170c3f54e16dcba3e9c5f1d516a45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:20:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/857fcac51e3d4f168abf8df8b605a646\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a6cd2b5f344538baa58f77d4170c91 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7744650d02062b99115535fb02108b6eeab6f7b9","revisions":{"d38ed3431edf1de5d6e6530ea9aa09cacb1725a3":{"kind":"REWORK","_number":1,"created":"2025-06-18 22:04:39.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/952871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/952871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/1"}}},"commit":{"parents":[{"commit":"d518d76b48664ad470f2d9f2135d71a2de6d7503","subject":"Unpin isort","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d518d76b48664ad470f2d9f2135d71a2de6d7503"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"subject":"Improve Row stringification","message":"Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nChange-Id: I96ead59a0b278ab87a2548f7bff1252e1071dfbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d38ed3431edf1de5d6e6530ea9aa09cacb1725a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d38ed3431edf1de5d6e6530ea9aa09cacb1725a3"}]},"branch":"refs/heads/master"},"7334f2711cf6148c9dfb595d14fd4f124a696079":{"kind":"REWORK","_number":2,"created":"2025-06-18 22:08:04.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/952871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/952871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/2"}}},"commit":{"parents":[{"commit":"d518d76b48664ad470f2d9f2135d71a2de6d7503","subject":"Unpin isort","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d518d76b48664ad470f2d9f2135d71a2de6d7503"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 22:07:59.000000000","tz":-300},"subject":"Improve Row stringification","message":"Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nChange-Id: I96ead59a0b278ab87a2548f7bff1252e1071dfbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7334f2711cf6148c9dfb595d14fd4f124a696079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7334f2711cf6148c9dfb595d14fd4f124a696079"}]},"branch":"refs/heads/master"},"69ea0e67a3c0b77b76324547d30bee986af7e667":{"kind":"REWORK","_number":3,"created":"2025-10-02 02:48:45.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/952871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/952871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/3"}}},"commit":{"parents":[{"commit":"d28c9f9e5c00690ff0802d7f1ec1a8c3481234c0","subject":"Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d28c9f9e5c00690ff0802d7f1ec1a8c3481234c0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-10-02 02:48:37.000000000","tz":-300},"subject":"Improve Row stringification","message":"Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nChange-Id: I96ead59a0b278ab87a2548f7bff1252e1071dfbb\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/69ea0e67a3c0b77b76324547d30bee986af7e667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/69ea0e67a3c0b77b76324547d30bee986af7e667"}]},"branch":"refs/heads/master"},"6c5091ea8283db6411e6554fa28b50590ffcf01f":{"kind":"REWORK","_number":4,"created":"2025-10-02 05:04:53.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/952871/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/952871/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/4"}}},"commit":{"parents":[{"commit":"e810ebe65af9a7f926c521f7a353212e3f7acb71","subject":"Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e810ebe65af9a7f926c521f7a353212e3f7acb71"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-10-02 05:02:48.000000000","tz":-300},"subject":"Improve Row stringification","message":"Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nChange-Id: I96ead59a0b278ab87a2548f7bff1252e1071dfbb\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6c5091ea8283db6411e6554fa28b50590ffcf01f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6c5091ea8283db6411e6554fa28b50590ffcf01f"}]},"branch":"refs/heads/master"},"897a2f59d94b016ced66423f540dab45c8e638aa":{"kind":"REWORK","_number":5,"created":"2025-10-02 15:12:34.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/952871/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/952871/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/5"}}},"commit":{"parents":[{"commit":"e810ebe65af9a7f926c521f7a353212e3f7acb71","subject":"Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e810ebe65af9a7f926c521f7a353212e3f7acb71"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-10-02 15:12:22.000000000","tz":-300},"subject":"Improve Row stringification","message":"Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nChange-Id: I96ead59a0b278ab87a2548f7bff1252e1071dfbb\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/897a2f59d94b016ced66423f540dab45c8e638aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/897a2f59d94b016ced66423f540dab45c8e638aa"}]},"branch":"refs/heads/master"},"a53248df2c83a006db5c547eeddc8aeaa7f58a4d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-23 20:51:37.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/952871/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/952871/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/6"}}},"commit":{"parents":[{"commit":"76170f66a33775d05bb41dd04f5995da3bc01fc1","subject":"Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/76170f66a33775d05bb41dd04f5995da3bc01fc1"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-10-23 20:51:37.000000000","tz":0},"subject":"Improve Row stringification","message":"Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nChange-Id: I96ead59a0b278ab87a2548f7bff1252e1071dfbb\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a53248df2c83a006db5c547eeddc8aeaa7f58a4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a53248df2c83a006db5c547eeddc8aeaa7f58a4d"}]},"branch":"refs/heads/master","description":"Rebase"},"7744650d02062b99115535fb02108b6eeab6f7b9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-11-13 09:54:16.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"real_uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/71/952871/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/952871/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/952871/7"}}},"commit":{"parents":[{"commit":"9c1790796d8094d386257ebd2909186a9d63352c","subject":"Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9c1790796d8094d386257ebd2909186a9d63352c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-11-13 09:54:16.000000000","tz":0},"subject":"Improve Row stringification","message":"Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nChange-Id: I96ead59a0b278ab87a2548f7bff1252e1071dfbb\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7744650d02062b99115535fb02108b6eeab6f7b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7744650d02062b99115535fb02108b6eeab6f7b9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
