)]}'
{"id":"openstack%2Fovsdbapp~957271","triplet_id":"openstack%2Fovsdbapp~master~I561ddacb222d6346e79018a57d31037293ff5d67","project":"openstack/ovsdbapp","branch":"master","topic":"ha-chassis-group","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-02-17 21:12:57.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-02-17 21:12:57.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2025-12-23 20:43:52.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-02-17 21:12:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I561ddacb222d6346e79018a57d31037293ff5d67","subject":"LrpAddCommand can consume ha-chassis-group name","status":"MERGED","created":"2025-08-13 17:22:21.000000000","updated":"2026-02-17 21:13:47.000000000","submitted":"2026-02-17 21:12:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"957271-ha-chassis-group","meta_rev_id":"effb501185daca7a25cdcf82d2575ad9003f31f7","_number":957271,"virtual_id_number":957271,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 21:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2026-02-17 19:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-23 20:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2026-02-17 19:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-13 18:58:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-13 19:26:53.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2025-12-31 21:38:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-12-31 21:38:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-12-31 21:38:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-02-17 19:40:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"18c3d63134d27f54a889cc35c774d1df9f027f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-08-13 17:22:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cfe1aa87bc4ac13eca7d273d0b05c0518b1dfd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-08-13 17:24:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e9e40fb21a396e8ecffdddc4790c967b190641c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-08-13 17:37:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f8b023c773b95e76ed5f85d0c53c570cfc0d9592","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-08-13 17:42:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"53f9399d2086a737a8c6f69f9fc2f64baa3b31a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 18:58:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31fb400398a74789a86d1d23ef036c5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87afdad8a7e34810a1d2be0488fe7800 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d37078442354b82a1085dd3c82f4279 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/094aa306f0964179a5016c6e5044159a : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c14897b12e474125947a047e090c5c2e : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844b24da26ba4e0da599936b80f75215 : SUCCESS in 7m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/afd4796171ce4edeba87614104a8e81f : SUCCESS in 6m 40s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6a1c6d1b02024528b6cfa2f364ccee21 : SUCCESS in 1h 12m 45s","accounts_in_message":[],"_revision_number":4},{"id":"12204bb3e5f564223b46605a71cad5df48e22874","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-08-13 19:26:53.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4187e053bbbcbba0bb0ac50c5b7d438f186dadc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-08-13 19:31:12.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5723aaae85c2fe7fec2404398475d0897557418d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-08-13 19:31:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"426decc764f3988bb5e70ababa38f195d11f5108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-08-13 20:03:13.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f1e0658467db0681a0f830ef1ec5f993d9119d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 21:49:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41667b344fce4e12a1252d83fc7a6fee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ffcc45bd6174e91875f8188f7c76bae : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d55f89b6e8a4a849d33f78f1f7dfc01 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a74964aeb544dfbafdc2f6f50817e43 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2c852e14b884788850d5dd83598264f : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc78baf21964111bff69f3492d75c98 : SUCCESS in 6m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3ff11d8c899043c98097292f2bc16692 : SUCCESS in 5m 49s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1631e76effeb43eba5a9fbf2635e103c : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":6},{"id":"467dac3c8041ae0fbbf0ecf37c0c1cd6891b107e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-12-23 20:43:52.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b990c4193004c20ef1ebf5e75827b610fea816ef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-17 19:40:49.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"8d1d0b50dd5ea2ae113a73fc8408137c81406ef2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:41:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"aa33f1365359a7924354e8e026f6d88647330667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 21:12:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/565f628a011f4f2ba5e97cf37ac5c4f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52f666130f164236a665c4dea81cd335 : SUCCESS in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1bd386e0f844fbf8ddcedce3344757e : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f07c7dacf39942d293fc86a8353434ae : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e39672eb0d4af8b6d03e12aa8d3bcb : SUCCESS in 6m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/83e1413a99d449ac8b7640df245b7f17 : SUCCESS in 5m 41s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/22d2200c1ac74a48ac9f66b8bac75429 : SUCCESS in 1h 27m 14s","accounts_in_message":[],"_revision_number":6},{"id":"b8cd3ec74bf01e7e3a818043aee1ea3eb4a591a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 21:12:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"effb501185daca7a25cdcf82d2575ad9003f31f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 21:13:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbd23da502c4e6eae0a4c5ac372d624\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28fb97d7056d48cc8518ee58a21bdf37 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"eadd3ac1324b751300e4b0a37104c0aed268414a","revisions":{"702805f3417763bbcf752aec571258af97579935":{"kind":"REWORK","_number":1,"created":"2025-08-13 17:22:21.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/71/957271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/957271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/1"}}},"commit":{"parents":[{"commit":"085ee75dee277839c04e54535d06263aebfed776","subject":"Merge \"nb: add support for chains in logical router policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/085ee75dee277839c04e54535d06263aebfed776"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:20:48.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:20:48.000000000","tz":-240},"subject":"LrpAddCommand can consume ha-chassis-group name","message":"LrpAddCommand can consume ha-chassis-group name\n\nSimilarly to setting gateway_chassis, the ha_chassis_group column can be\nset using the ha_chassis_group name and not just its uuid.\n\nChange-Id: I561ddacb222d6346e79018a57d31037293ff5d67\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nAssisted-By: claude-4-sonnet\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/702805f3417763bbcf752aec571258af97579935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/702805f3417763bbcf752aec571258af97579935"}]},"branch":"refs/heads/master"},"eee9c0f82b8321de284063194177f42bc5786df0":{"kind":"REWORK","_number":2,"created":"2025-08-13 17:24:38.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/71/957271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/957271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/2"}}},"commit":{"parents":[{"commit":"085ee75dee277839c04e54535d06263aebfed776","subject":"Merge \"nb: add support for chains in logical router policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/085ee75dee277839c04e54535d06263aebfed776"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:20:48.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:24:20.000000000","tz":-240},"subject":"LrpAddCommand can consume ha-chassis-group name","message":"LrpAddCommand can consume ha-chassis-group name\n\nSimilarly to setting gateway_chassis, the ha_chassis_group column can be\nset using the ha_chassis_group name and not just its uuid.\n\nChange-Id: I561ddacb222d6346e79018a57d31037293ff5d67\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nAssisted-By: claude-4-sonnet\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eee9c0f82b8321de284063194177f42bc5786df0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eee9c0f82b8321de284063194177f42bc5786df0"}]},"branch":"refs/heads/master"},"33864b5ba446155019ced6d30716d9619933b70c":{"kind":"REWORK","_number":3,"created":"2025-08-13 17:37:41.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/71/957271/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/957271/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/3"}}},"commit":{"parents":[{"commit":"085ee75dee277839c04e54535d06263aebfed776","subject":"Merge \"nb: add support for chains in logical router policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/085ee75dee277839c04e54535d06263aebfed776"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:20:48.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:37:36.000000000","tz":-240},"subject":"LrpAddCommand can consume ha-chassis-group name","message":"LrpAddCommand can consume ha-chassis-group name\n\nSimilarly to setting gateway_chassis, the ha_chassis_group column can be\nset using the ha_chassis_group name and not just its uuid.\n\nChange-Id: I561ddacb222d6346e79018a57d31037293ff5d67\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nAssisted-By: claude-4-sonnet\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/33864b5ba446155019ced6d30716d9619933b70c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/33864b5ba446155019ced6d30716d9619933b70c"}]},"branch":"refs/heads/master"},"e7cd9f77bea4a4a5e09cac4a4f27be889c376ccb":{"kind":"REWORK","_number":4,"created":"2025-08-13 17:42:03.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/71/957271/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/957271/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/4"}}},"commit":{"parents":[{"commit":"085ee75dee277839c04e54535d06263aebfed776","subject":"Merge \"nb: add support for chains in logical router policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/085ee75dee277839c04e54535d06263aebfed776"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:20:48.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:41:47.000000000","tz":-240},"subject":"LrpAddCommand can consume ha-chassis-group name","message":"LrpAddCommand can consume ha-chassis-group name\n\nSimilarly to setting gateway_chassis, the ha_chassis_group column can be\nset using the ha_chassis_group name and not just its uuid.\n\nChange-Id: I561ddacb222d6346e79018a57d31037293ff5d67\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nAssisted-By: claude-4-sonnet\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e7cd9f77bea4a4a5e09cac4a4f27be889c376ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e7cd9f77bea4a4a5e09cac4a4f27be889c376ccb"}]},"branch":"refs/heads/master"},"1dddc6992c141393504bc0cb31fac410ca1079de":{"kind":"REWORK","_number":5,"created":"2025-08-13 19:31:12.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/71/957271/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/957271/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/5"}}},"commit":{"parents":[{"commit":"085ee75dee277839c04e54535d06263aebfed776","subject":"Merge \"nb: add support for chains in logical router policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/085ee75dee277839c04e54535d06263aebfed776"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:20:48.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 19:31:12.000000000","tz":0},"subject":"LrpAddCommand can consume ha-chassis-group name","message":"LrpAddCommand can consume ha-chassis-group name\n\nSimilarly to setting gateway_chassis, the ha_chassis_group column can be\nset using the ha_chassis_group name and not just its uuid.\n\nChange-Id: I561ddacb222d6346e79018a57d31037293ff5d67\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nAssisted-By: claude-4-sonnet\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1dddc6992c141393504bc0cb31fac410ca1079de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1dddc6992c141393504bc0cb31fac410ca1079de"}]},"branch":"refs/heads/master"},"eadd3ac1324b751300e4b0a37104c0aed268414a":{"kind":"NO_CHANGE","_number":6,"created":"2025-08-13 20:03:13.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/71/957271/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/71/957271/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/71/957271/6"}}},"commit":{"parents":[{"commit":"085ee75dee277839c04e54535d06263aebfed776","subject":"Merge \"nb: add support for chains in logical router policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/085ee75dee277839c04e54535d06263aebfed776"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 17:20:48.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-08-13 19:49:35.000000000","tz":-240},"subject":"LrpAddCommand can consume ha-chassis-group name","message":"LrpAddCommand can consume ha-chassis-group name\n\nSimilarly to setting gateway_chassis, the ha_chassis_group column can be\nset using the ha_chassis_group name and not just its uuid.\n\nChange-Id: I561ddacb222d6346e79018a57d31037293ff5d67\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nAssisted-By: claude-4-sonnet\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eadd3ac1324b751300e4b0a37104c0aed268414a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eadd3ac1324b751300e4b0a37104c0aed268414a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
