)]}'
{"id":"openstack%2Fovsdbapp~965352","triplet_id":"openstack%2Fovsdbapp~master~I903334f4566ac6dd28452a21cde5ae96fd9a26e3","project":"openstack/ovsdbapp","branch":"master","topic":"get_uuid_rowview_row","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-11-13 23:19:58.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2025-11-13 23:19:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I903334f4566ac6dd28452a21cde5ae96fd9a26e3","subject":"Improve ``get_uuid``, now accepts RowView|Row","status":"MERGED","created":"2025-10-30 09:10:34.000000000","updated":"2025-11-13 23:20:46.000000000","submitted":"2025-11-13 23:19:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965352-get_uuid_rowview_row","meta_rev_id":"0988fad1a8efc6fa625680b74b0d5dd097e40210","_number":965352,"virtual_id_number":965352,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-13 23:19:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-11-13 21:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2025-11-13 14:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-11-13 21:49:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-30 09:11:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2025-10-30 10:04:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-10 16:43:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-11-13 14:04:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-11-13 21:49:30.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"7a202adb97b806974df103f268a311dd5d70d114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-30 09:10:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29804925ef1f108255331c34329e1767017a9d0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 10:04:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7fb1410ac5646ea81e009472a6dfb5e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515ed14f612b4409a50dea8ccb90d831 : FAILURE in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f362f92cbfa64cb99d5057cc7a15cc91 : SUCCESS in 2m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17cc575b05ad46b6ad1c9392c50fa928 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f820b684f14870974229f58db2dad7 : SUCCESS in 3m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/538b76f9f4e044278afb9fa1662dbfb7 : SUCCESS in 6m 47s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/096eb958ffeb48969edeb50acc9c97e1 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f8c58d39702fca7d280eb6e1d57eca18192593f0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-10 16:43:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f469bc2784c7245b616ab37158e401ab3b9d66f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 17:55:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e9687103d9a4f5d9e1b107948594b63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20dbe7421de841cdb6c0921acf77d9b7 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47c22ae825c34713b5428004af7f87a2 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f696839df5ce4d15ace9f93471424cab : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ee10ab303240b7af3433c51e86921a : SUCCESS in 5m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9567c3b215a341afa07ca924ffde5037 : SUCCESS in 7m 47s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8a3d87554e9049289531a0437bd77206 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e3eb16f090b93240b4174ee0625c547d1e9fa326","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-11-11 17:01:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b29009820cf216844532be27b0465cdd82a040a4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-11-12 18:56:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bf75b86f12170f7f0cfd30142f8b083cca44c1b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-13 09:54:57.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18208052ceb70ddbb96e43ea34e70abbdc27bfee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-13 10:36:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f05e05326d7331448ab4907dd593557b7fcb5429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 11:22:35.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2cced81a3a443a895129ccef2951e58\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/386e52ce1ea74acd941f829c91f5d241 : FAILURE in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ac42b98a2bf44de9fc6e4625e5859ce : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/656d49e1645a471d8742f3a6b9d1ebf7 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d39dee258844909530cfa2649cf293 : SUCCESS in 5m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/51f0eb77f16c47bab2bf9f3a4d00e02d : SUCCESS in 6m 36s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7b00cb23fe644912b619d691ad1ac6b1 : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":2},{"id":"8737377b68ed7247033de886ee56bb3d5045ef88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-13 11:26:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9e3312b0017bed7619f6b2010b002c77136b1be6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 12:11:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59f503065caa4ec9b410ffb338dd09f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a44876104014874958ebb5238347398 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7333e22bc990416497f2ba1ca495b88a : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ce7a2db0e1447da80028718a2a35823 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed956bfd0ee2402cb08ef63cae89731e : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/615ec625156e46de99c77f37b6d0e7d7 : SUCCESS in 4m 51s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dce78507838f4b9d91a8e81b5790f45e : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":3},{"id":"f575b84ae479f429b03c6f0f17f2855a4c3cd01f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-13 14:04:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"177054955994710f8a7b9070b1e6e19cd7793e47","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2025-11-13 21:49:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"71be8d2ca44a7211dec80388928bc7a38e288b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 21:49:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"95471ef071a4d4ddf3c09f8a47527b94803defbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 23:19:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b827fb11e9234730b72ffc6b2986db69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b354fb28464cbe8c263d8a39ec05f5 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e00505d6c4724e288c9275f9f13674f9 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8806c879bcf448f8ab55d4d833530b2 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326a497075954d8c87ac0834fe933e72 : SUCCESS in 6m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/05d84aaf8f8c4bfe91a236e8604c15ea : SUCCESS in 4m 37s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e5d797968288446ebf81a5dba9289aa1 : SUCCESS in 1h 18m 09s","accounts_in_message":[],"_revision_number":3},{"id":"a06feceb3eeb06c033f72b00038db5faa8d9aba8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 23:19:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0988fad1a8efc6fa625680b74b0d5dd097e40210","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 23:20:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fdf55403cb84ed09391ce9d9269881d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018463b1979e48d99b37473450f2642c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e69b710788edd6fee1cc4acbd7ef0ad34a9b97ca","revisions":{"1a53d2b5de0eb26e05d7ac4dd9aa997ecf3c6558":{"kind":"REWORK","_number":1,"created":"2025-10-30 09:10:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/965352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/965352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/1"}}},"commit":{"parents":[{"commit":"856acf3f422f8862e17c9cb5c1245b767c24d06c","subject":"Merge \"Do not force oslo\u0027s jsonutils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/856acf3f422f8862e17c9cb5c1245b767c24d06c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-30 09:08:52.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-30 09:10:13.000000000","tz":0},"subject":"Improve ``get_uuid``, accepts RowView|Row","message":"Improve ``get_uuid``, accepts RowView|Row\n\nThe methdo ``get_uuid`` now accepts ``RowView`` and ``Row`` input\nvalues. The UUID field is a member of these classes.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I903334f4566ac6dd28452a21cde5ae96fd9a26e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1a53d2b5de0eb26e05d7ac4dd9aa997ecf3c6558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1a53d2b5de0eb26e05d7ac4dd9aa997ecf3c6558"}]},"branch":"refs/heads/master"},"2ae3ebccbdfd3c3bffb8ce5cae3d7cb59053b6a4":{"kind":"REWORK","_number":2,"created":"2025-11-13 10:36:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/965352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/965352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/2"}}},"commit":{"parents":[{"commit":"a30b69b16005338d51a119912a9ad2b80544bc9f","subject":"Merge \"reno: Update master for unmaintained/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a30b69b16005338d51a119912a9ad2b80544bc9f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-30 09:08:52.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-13 10:33:53.000000000","tz":0},"subject":"Improve ``get_uuid``, now accepts RowView|Row","message":"Improve ``get_uuid``, now accepts RowView|Row\n\nThe methdo ``get_uuid`` now accepts ``RowView`` and ``Row`` input\nvalues. The UUID field is a member of these classes.\n\nIn general, the ``get_uuid`` method retrieves the object ``uuid`` field,\nif present, or returns the tested object itself. In case of\n``BaseCommand`` derived objects, the method first retrieves the\n``result`` value, that should store a ``RowView|Row`` class object.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I903334f4566ac6dd28452a21cde5ae96fd9a26e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ae3ebccbdfd3c3bffb8ce5cae3d7cb59053b6a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ae3ebccbdfd3c3bffb8ce5cae3d7cb59053b6a4"}]},"branch":"refs/heads/master"},"e69b710788edd6fee1cc4acbd7ef0ad34a9b97ca":{"kind":"REWORK","_number":3,"created":"2025-11-13 11:26:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/965352/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/52/965352/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/52/965352/3"}}},"commit":{"parents":[{"commit":"a30b69b16005338d51a119912a9ad2b80544bc9f","subject":"Merge \"reno: Update master for unmaintained/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a30b69b16005338d51a119912a9ad2b80544bc9f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-30 09:08:52.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-13 11:25:47.000000000","tz":0},"subject":"Improve ``get_uuid``, now accepts RowView|Row","message":"Improve ``get_uuid``, now accepts RowView|Row\n\nThe methdo ``get_uuid`` now accepts ``RowView`` and ``Row`` input\nvalues. The UUID field is a member of these classes.\n\nIn general, the ``get_uuid`` method retrieves the object ``uuid`` field,\nif present, or returns the tested object itself. In case of\n``BaseCommand`` derived objects, the method first retrieves the\n``result`` value, that should store a ``RowView|Row`` class object.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I903334f4566ac6dd28452a21cde5ae96fd9a26e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e69b710788edd6fee1cc4acbd7ef0ad34a9b97ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e69b710788edd6fee1cc4acbd7ef0ad34a9b97ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
