)]}'
{"id":"openstack%2Fovsdbapp~968031","triplet_id":"openstack%2Fovsdbapp~master~Ie29883a10a7ccfbf87dc8f835f109e771f256584","project":"openstack/ovsdbapp","branch":"master","topic":"neutron-functional","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-28 13:29:23.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-04-28 12:18:22.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-02-18 11:30:44.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-04-28 13:29:23.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-04-28 12:19:46.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"Ie29883a10a7ccfbf87dc8f835f109e771f256584","subject":"Add Neutron functional test to the gate queue","status":"MERGED","created":"2025-11-21 15:23:13.000000000","updated":"2026-04-28 13:32:48.000000000","submitted":"2026-04-28 13:29:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968031-neutron-functional","meta_rev_id":"db0d8457130459cd6d46b08d311d4d02b0217ad8","_number":968031,"virtual_id_number":968031,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-28 13:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-04-28 12:18:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2026-04-28 12:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-04-28 12:18:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-28 09:28:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-21 16:33:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-11-21 16:35:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2025-11-21 16:35:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-11-21 16:35:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2025-11-21 16:54:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 08:25:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-11-24 09:00:50.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-02-17 19:42:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-04-16 13:09:57.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"4fec05c74650bde195470ed4da77db26afc3fbde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-21 15:23:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99ef1177152279cdbd3d02c0581e21e40fc8a0f1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-21 16:33:18.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"e023d67a33f35cc19437550c5d5f834dfec5488e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 16:54:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a250e5ff821b47c39b7d1efcee1f03be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/173d7f9edd314f7e9dd9184e561a18f4 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cb6d1f90865406f992f3bc10dfed838 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32258ffa51ad4b55afe03f3877fc0a11 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ccec5481b1341558d41b1ea5ee0a3d6 : SUCCESS in 6m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/30fc08eebfb945ca9297fb474da3ae06 : SUCCESS in 5m 05s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1d1b233e65a641b093928691c4dea657 : SUCCESS in 1h 25m 09s","accounts_in_message":[],"_revision_number":1},{"id":"fb8e6276a21754cba6aff90ef2c98cbb47c4b372","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-24 08:25:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfcd18afe9a20e377adad6a1705464daf960f1fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 08:26:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5028da4f4aa4d3b5b3fc03fb5b026dbeace286d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-24 09:00:50.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f48c537499574be28d882ecf9c1ab896103d72d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 09:00:52.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"b2c7ad9b318b22616f355127b4416007b317f85d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-24 09:15:53.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"ff9532d0fcc9613c8818cdac8c4b42fb45711147","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-24 09:27:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d781d17a213b178efd2e12b99951f176de4d8c46","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-24 14:58:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9a76b8aa35d75643cb9485a40a9a4344c555b15","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-24 15:44:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"790d0f7886fd4b326647bdc6aa134c245bfb221f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-25 12:14:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6227c90aa4279b0d3f3d86814cb9f7520437f4da","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-25 12:20:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96afe55d9a8b967e42925efed16e6dc8e1f09ca8","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-25 21:03:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30f4a383114cbc9022130804b5420251ee9b8787","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-26 04:56:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d1400c09c8e2346a0c820bdb42479e41fd84dc0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 14:14:15.000000000","message":"Patch Set 1: -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18a205bb2caffe3d42ce3af3f65af3df41b40286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-14 22:35:02.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1, Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f85c21a75a6bac36660b0f8229610ac8e77f7206","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-14 22:35:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c99b0bc2a0a9e9d2a1ff6d58352b4149b691223f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 23:35:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb6ab72791864325a65a0e03a46c257d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ac8c052a9241368d38c0b985709229 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2839241b16944a9a84f536a599e7ca59 : SUCCESS in 2m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4e858c9aa6d49aea15c3b03264338c4 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e56cc04d2a458a9090e02d5ecff58d : SUCCESS in 3m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/095ae4284f05436daaaa189713db5e56 : SUCCESS in 8m 10s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b980c9b2837f4f2e9e348887467eb15f : SUCCESS in 59m 18s\n- neutron-functional-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/386b7aae6f254b2785b00d6d84b2daea : FAILURE in 16m 43s","accounts_in_message":[],"_revision_number":2},{"id":"7866e4f70fb3820ac5811e1dbe79d025e3c97cb2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-15 07:00:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b280a0f12f154d0b1dd11e4482bc468601076c4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-15 07:14:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a584421c296ce7e6312b4fe43b13800984a7d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 08:18:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6288cabd27e8474c9ead079c10ee3495\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2555083f4654b938b963bebb276325f : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5344397c41ea463c8ee5aca81149b544 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee0e5bd9978547f9b812ca6952a10929 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63712b00c52f48e787b8128d927e1029 : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8fece416852741589ff3fa73a98d7792 : SUCCESS in 5m 00s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e27ef81226dc4f2e8dc918ccef2f4dcf : SUCCESS in 1h 12m 01s\n- neutron-functional-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/27e2453b5a28480ba2c9454f56b5434d : FAILURE in 17m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fa6b596fd9e34e4e37b26d85e4b9c32d48a44005","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 07:17:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1755280506dd9700ae4d7553923f7d5f867f3130","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 07:17:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dedc81fba36f5e11789d8cc2deca347a6d8dd4b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 08:17:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f5a357e02414bb8a7b2d5eedb71eef7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e024ed5ac25246d287bd0726a441aee0 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/451d24ee450143c897fe56430d3de9ee : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a7cacf842e54a40bf4d85414ee528b1 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954da202492145b59ab5d6217247c7f3 : SUCCESS in 6m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4a59956e44824efcb2ba8763ddb09b55 : SUCCESS in 4m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1d204fecf0354e708d1fc2487871dc93 : SUCCESS in 58m 21s\n- ovsdbapp-neutron-functional-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/c5726748bebf49ef9504ac7d0748ced9 : FAILURE in 9m 08s","accounts_in_message":[],"_revision_number":3},{"id":"b19d8dbc158b3fb56377e11ae1be83f17a529a0b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-17 19:42:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39517003c3825928527d370a40c7e44dfd39ffcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 20:27:57.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea44f1b028d442d28b2bed712e9a0576\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d123a288374418ba7534b5c73be789 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07186dfdee794c1abc92020a1ac31552 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1b9db9f7c7e4f8a851c409d332e9882 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee403bf6fe034629b0314fe00186ee42 : SUCCESS in 6m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/220eaadfe54e42309cb8a99150ca83b2 : SUCCESS in 6m 31s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9e72b894476f4d1da1841810c32754ee : SUCCESS in 42m 02s\n- ovsdbapp-neutron-functional-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/36f333e583d8409d97cc14c4ffbba0b8 : FAILURE in 21m 45s","accounts_in_message":[],"_revision_number":3},{"id":"5112902b85a6946ab488c31f5b66fdd75671e4b9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-18 11:30:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bc1520863d0110a812a019d09fed36c0db0f285d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-12 11:13:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad59e13fe90b80ec6e502500be1c92fbbb1795aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-15 10:58:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f9570fea7d7b88b195a942852ccfb6296678508f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 11:00:00.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/ovsdbapp on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][2][\u0027neutron-functional\u0027]\n\nThe problem appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n      check:\n        jobs:\n          - openstack-tox-functional\n          - ovsdbapp-neutron-ovn-tempest-ovs-release\n  ...\n\n  in \"openstack/ovsdbapp/zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":4},{"id":"b53d7dc0d0f24ee1dee181751dc32070a04279b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-15 11:03:43.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"20a7938143986a360a2eb7aad39b659f6f05d6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 11:04:04.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/ovsdbapp on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][2][\u0027neutron-functional\u0027]\n\nThe problem appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n      check:\n        jobs:\n          - openstack-tox-functional\n          - ovsdbapp-neutron-ovn-tempest-ovs-release\n  ...\n\n  in \"openstack/ovsdbapp/zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":5},{"id":"aa79a9d584d3883fc2cf9cdf3287d52d31beace7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-15 11:05:25.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"6679e43605064b63a667622e30aefa93375a239a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-15 11:13:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"befb1658bc8dc44bb79c55b1ecaa92e116a05b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 12:37:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4981c932b0174d5ebe5acfea21f11ea7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf30c80e485447d2a7609fbd79e21b47 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9e0251590734d2cb19e3665809f9ca9 : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab9900ebc21c4af0bc0176784dbd8a37 : SUCCESS in 1m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1cfc70b6576a44a888d780e00e92c5e3 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23582449700d4667a17b314e98add915 : SUCCESS in 6m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7bacadcd4fc84effbfcde61fbbc2a644 : SUCCESS in 8m 23s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/be173a705df7403cbf6cb8e57020db4b : SUCCESS in 43m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7dd2ee3e6f5e414f8f7f147b41227b33 : SUCCESS in 1h 25m 34s","accounts_in_message":[],"_revision_number":6},{"id":"1daf1aedc82a9199b81ee249995a59624caaf147","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-15 13:04:58.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"52c7a2c9f89a2e373978e15e7f9ce8371a3fbb07","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-16 05:30:38.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"b6a356e3a6e6f7cd318a8209a3b1a15e0d76dfc6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-04-16 13:09:57.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"648bea4328a84887e898dde66c12c49077e58f4d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-23 15:06:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bfc851351c1638ed2fdd40fb9897d6b28bf67db9","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-23 15:08:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c67183425427b0cdc2bdaa28e11ecc4ec72a2387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-23 15:30:52.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"6d0342cdd2e780835500ab663f4d3b3bf5aba824","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-23 15:33:38.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8715429cefd59c67ac8b84396d5c0d7902bbb9d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:40:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/363605b1510846fcb5c2eb4bf4c8b92e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6efd78f30c4754a0bb239e6c330870 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1ed298cdc784cd89a7a797a97e2ea52 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14fbe9ef761c49108650931a6633bf82 : SUCCESS in 2m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f13d4df93e8b4c3b87da19799e76c415 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e61fd0518cf47eda3283159018e544b : SUCCESS in 7m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4423aa00722b48f7aa58058055267474 : SUCCESS in 5m 56s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c1be9c4941984dc7a8001c94999ddf78 : SUCCESS in 1h 07m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0bdf758419274cf2bd70a34f808e690a : FAILURE in 18m 49s","accounts_in_message":[],"_revision_number":7},{"id":"e4b12918a1793397023408b3f9d56adf4d986530","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-28 08:56:05.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f0a7617a6e97069104318c798de4f72297edef0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-28 08:56:24.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"e55092fe53072f7cd4db8f8f40c9e5b5b4cc8233","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-28 09:28:00.000000000","message":"Patch Set 8: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8acf87614c389b6590187d0bc501e0e4665317cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 10:22:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f09d6877a7eb4c988876eb87154ce6ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531aa95f6cc649d2852066bfb536555d : SUCCESS in 2m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8593ced564747f69ac0f9dc1086333b : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/413786db666f481981bb0e9e6d544ab2 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c24114e2b8ea463d834089a575c1c7c9 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f38bc726db074d278ceebc1548532227 : SUCCESS in 6m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ba27fd6b51134601aa02c94bb2e430d0 : SUCCESS in 4m 21s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/931d83f9af75463e9e020d4e49420838 : SUCCESS in 1h 20m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2c4a3ae034fd4874bac945672a7b9448 : SUCCESS in 46m 31s","accounts_in_message":[],"_revision_number":8},{"id":"dcd32db8ca447a0c7c0fb0322fce43136fe23494","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-28 12:08:46.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"02267d15f067a2bb8a73ed5c998195a3e8728475","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-28 12:09:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"acfb9918a9a1e15a98fd0a9201dab28e5e393684","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-04-28 12:18:22.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6c82ed61e2693debe4c28b6368c776e92dfcc7b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 12:18:59.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"cc2af24b88f8d6705293e94947e9af66473c348e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-28 12:19:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c4e87f9b8cd1568c71803c295924ef2a2abeb85f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 13:29:23.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0813d170c71b439bae007a5b45bd797c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7daa413a06274230b66a0b53d3bbfeb5 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fff7bc708a64842b18245d045182dbf : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d9bb09c6ef940eab76b968a4fb7aa31 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba337b17bba4a8ab1019a6152fbd6c2 : SUCCESS in 4m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/375a5e4f3bc24bdd9133a62f7195d2e2 : SUCCESS in 5m 36s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c0633edb7c194c4e88613df80f1005e3 : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":8},{"id":"33ddeb5d9ca7ff41d0c99cfff67d3064fbbf7ecf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 13:29:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"db0d8457130459cd6d46b08d311d4d02b0217ad8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 13:32:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/630b332303454a6480d2d96fc49802e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b193db17f514bbc9b1eea2599c5d3fd : SUCCESS in 51s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e5fe03382e59c8b1df0442f4432d1b2bf22bb53d","revisions":{"8da340842c1eba46be83b486b2ea908a3a1f6591":{"kind":"REWORK","_number":1,"created":"2025-11-21 15:23:13.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/31/968031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/968031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/1"}}},"commit":{"parents":[{"commit":"61cde4703cf032c7ce6bd8dce10c5822dbe12b09","subject":"Revert \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cde4703cf032c7ce6bd8dce10c5822dbe12b09"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-11-21 15:21:39.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-11-21 15:23:03.000000000","tz":-300},"subject":"Add Neutron functional test to the gate queue","message":"Add Neutron functional test to the gate queue\n\nThere was a recent issue with ovsdbapp [1] that broke the Neutron gate.\nThis patch proposes adding a Neutron functional job to the gate queue\nto catch such issues earlier next time and avoid breaking the gate.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2131994\n\nChange-Id: Ie29883a10a7ccfbf87dc8f835f109e771f256584\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8da340842c1eba46be83b486b2ea908a3a1f6591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8da340842c1eba46be83b486b2ea908a3a1f6591"}]},"branch":"refs/heads/master"},"92cdd3d987083e92dba839e06ceeea8abb6ff683":{"kind":"REWORK","_number":2,"created":"2026-01-14 22:35:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/968031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/968031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/2"}}},"commit":{"parents":[{"commit":"61cde4703cf032c7ce6bd8dce10c5822dbe12b09","subject":"Revert \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cde4703cf032c7ce6bd8dce10c5822dbe12b09"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-11-21 15:21:39.000000000","tz":-300},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-01-14 22:35:02.000000000","tz":0},"subject":"Add Neutron functional test to the gate queue","message":"Add Neutron functional test to the gate queue\n\nThere was a recent issue with ovsdbapp [1] that broke the Neutron gate.\nThis patch proposes adding a Neutron functional job to the gate queue\nto catch such issues earlier next time and avoid breaking the gate.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2131994\n\nChange-Id: Ie29883a10a7ccfbf87dc8f835f109e771f256584\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/92cdd3d987083e92dba839e06ceeea8abb6ff683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/92cdd3d987083e92dba839e06ceeea8abb6ff683"}]},"branch":"refs/heads/master"},"6c07877a442eee11859c97b52907de79da5cd14f":{"kind":"REWORK","_number":3,"created":"2026-01-16 07:17:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/968031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/968031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/3"}}},"commit":{"parents":[{"commit":"b4b46458362e299682bbbd23016c29f8e5c0d761","subject":"windows: Add close() method to WaitQueue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b4b46458362e299682bbbd23016c29f8e5c0d761"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-11-21 15:21:39.000000000","tz":-300},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-16 07:17:21.000000000","tz":60},"subject":"Add Neutron functional test to the gate queue","message":"Add Neutron functional test to the gate queue\n\nThere was a recent issue with ovsdbapp [1] that broke the Neutron gate.\nThis patch proposes adding a Neutron functional job to the gate queue\nto catch such issues earlier next time and avoid breaking the gate.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2131994\n\nChange-Id: Ie29883a10a7ccfbf87dc8f835f109e771f256584\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6c07877a442eee11859c97b52907de79da5cd14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6c07877a442eee11859c97b52907de79da5cd14f"}]},"branch":"refs/heads/master"},"9228ddc0423a8b04773ec13078fe89a4701cb683":{"kind":"REWORK","_number":4,"created":"2026-04-15 10:58:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/968031/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/968031/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/4"}}},"commit":{"parents":[{"commit":"2d63f81ad7664c1b7038474671aa6ed5cf4dd8d5","subject":"Merge \"venv: Stop OVN ovsdb-server processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2d63f81ad7664c1b7038474671aa6ed5cf4dd8d5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-15 10:57:17.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-15 10:58:22.000000000","tz":120},"subject":"Add Neutron functional test to the gate queue","message":"Add Neutron functional test to the gate queue\n\nThere was a recent issue with ovsdbapp [1] that broke the Neutron gate.\nThis patch proposes adding a Neutron functional job to the gate queue\nto catch such issues earlier next time and avoid breaking the gate.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2131994\n\nChange-Id: Ie29883a10a7ccfbf87dc8f835f109e771f256584\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9228ddc0423a8b04773ec13078fe89a4701cb683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9228ddc0423a8b04773ec13078fe89a4701cb683"}]},"branch":"refs/heads/master"},"c81d4e5d435912ad7e94d0d470df0e95f251d2ac":{"kind":"REWORK","_number":5,"created":"2026-04-15 11:03:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/968031/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/968031/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/5"}}},"commit":{"parents":[{"commit":"2d63f81ad7664c1b7038474671aa6ed5cf4dd8d5","subject":"Merge \"venv: Stop OVN ovsdb-server processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2d63f81ad7664c1b7038474671aa6ed5cf4dd8d5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-15 10:57:17.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-15 11:03:37.000000000","tz":120},"subject":"Add Neutron functional test to the gate queue","message":"Add Neutron functional test to the gate queue\n\nThere was a recent issue with ovsdbapp [1] that broke the Neutron gate.\nThis patch proposes adding a Neutron functional job to the gate queue\nto catch such issues earlier next time and avoid breaking the gate.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2131994\n\nChange-Id: Ie29883a10a7ccfbf87dc8f835f109e771f256584\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c81d4e5d435912ad7e94d0d470df0e95f251d2ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c81d4e5d435912ad7e94d0d470df0e95f251d2ac"}]},"branch":"refs/heads/master"},"f6a527c3fc1b53e094bebae326eb949f4828c815":{"kind":"REWORK","_number":6,"created":"2026-04-15 11:05:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/968031/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/968031/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/6"}}},"commit":{"parents":[{"commit":"2d63f81ad7664c1b7038474671aa6ed5cf4dd8d5","subject":"Merge \"venv: Stop OVN ovsdb-server processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2d63f81ad7664c1b7038474671aa6ed5cf4dd8d5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-15 10:57:17.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-15 11:05:19.000000000","tz":120},"subject":"Add Neutron functional test to the gate queue","message":"Add Neutron functional test to the gate queue\n\nThere was a recent issue with ovsdbapp [1] that broke the Neutron gate.\nThis patch proposes adding a Neutron functional job to the gate queue\nto catch such issues earlier next time and avoid breaking the gate.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2131994\n\nChange-Id: Ie29883a10a7ccfbf87dc8f835f109e771f256584\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f6a527c3fc1b53e094bebae326eb949f4828c815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f6a527c3fc1b53e094bebae326eb949f4828c815"}]},"branch":"refs/heads/master"},"2a41c7526db6d2a8c32aa118ba9c45917e452aff":{"kind":"REWORK","_number":7,"created":"2026-04-23 15:30:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/968031/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/968031/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/7"}}},"commit":{"parents":[{"commit":"3480ad11a731e0399769046ade12c4688c6bcadc","subject":"Merge \"Change signature of ``lsp_detach_mirror`` method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3480ad11a731e0399769046ade12c4688c6bcadc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-15 10:57:17.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-22 11:52:54.000000000","tz":120},"subject":"Add Neutron functional test to the gate queue","message":"Add Neutron functional test to the gate queue\n\nThere was a recent issue with ovsdbapp [1] that broke the Neutron gate.\nThis patch proposes adding a Neutron functional job to the gate queue\nto catch such issues earlier next time and avoid breaking the gate.\n\nThe test list has been reduced to the ML2/OVN plugin test cases.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2131994\n\nChange-Id: Ie29883a10a7ccfbf87dc8f835f109e771f256584\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2a41c7526db6d2a8c32aa118ba9c45917e452aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2a41c7526db6d2a8c32aa118ba9c45917e452aff"}]},"branch":"refs/heads/master"},"e5fe03382e59c8b1df0442f4432d1b2bf22bb53d":{"kind":"REWORK","_number":8,"created":"2026-04-28 08:56:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/968031/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/968031/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/968031/8"}}},"commit":{"parents":[{"commit":"3480ad11a731e0399769046ade12c4688c6bcadc","subject":"Merge \"Change signature of ``lsp_detach_mirror`` method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3480ad11a731e0399769046ade12c4688c6bcadc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-15 10:57:17.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-04-28 08:56:10.000000000","tz":120},"subject":"Add Neutron functional test to the gate queue","message":"Add Neutron functional test to the gate queue\n\nThere was a recent issue with ovsdbapp [1] that broke the Neutron gate.\nThis patch proposes adding a Neutron functional job to the gate queue\nto catch such issues earlier next time and avoid breaking the gate.\n\nThe test list has been reduced to the ML2/OVN plugin test cases.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2131994\n\nChange-Id: Ie29883a10a7ccfbf87dc8f835f109e771f256584\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e5fe03382e59c8b1df0442f4432d1b2bf22bb53d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e5fe03382e59c8b1df0442f4432d1b2bf22bb53d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
