)]}'
{"id":"openstack%2Fovsdbapp~973564","triplet_id":"openstack%2Fovsdbapp~master~Ic0c721fe088a53de98f1149e41ad5f72624be456","project":"openstack/ovsdbapp","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-28 18:39:17.000000000","reason":"Change was abandoned"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-01-28 18:38:59.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}},"hashtags":[],"change_id":"Ic0c721fe088a53de98f1149e41ad5f72624be456","subject":"Reapply \"Rework venv to support arbitrary schemas\"","status":"ABANDONED","created":"2026-01-15 22:54:10.000000000","updated":"2026-01-28 18:39:17.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc710d033a226113dc189a7198e6d04a641a7fbc","_number":973564,"virtual_id_number":973564,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 00:14:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 06:59:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-01-16 07:43:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b12c471659f5263db3db31db652c6f3a1b417abb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-15 22:54:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dac143df7a8c1ae1560cc93103ec2077b6052ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 00:14:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b95b1ceeb2a14b209fd0a175d071a7ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a7307853e34d639bbb4751bc0f5116 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1c7327c44da4522b0c1075c756f26c6 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/865d685510ff42a58c01f2d145fa6be5 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d5e1df927154ebbb1673da9d1bfe83a : SUCCESS in 6m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3b7ae7c1245449c89b8185742784d659 : SUCCESS in 6m 33s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ef77aa34400042fbb7b6409ff6660bd7 : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"31544d414b380f7d356ca655484e917d16dde6de","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 06:59:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be1be5131a657f20ff0355139fd01569921f84a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 07:43:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd2226e8965cd1e41e04a364582214a868cf9bca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-16 07:43:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab905b91c4101f69fa402fbb7bc02b0e166ef620","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-16 14:46:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcb6fe9a2f36c4838d1c2d9af77f6dc5c00f04c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-19 22:51:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"222522007588b3fc26db2a7b3049966bdbafa06a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-19 22:51:57.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9f8314afe54b933c2e12d249c08cce584ef41d80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 00:15:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3861e40788184d6599399351653b884c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5567f6154f914731bf542f6f33d8fdef : SUCCESS in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a404e0088e334a49ae9ba8ba402b0974 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72dd993533e94cacb6b9cc89dd2ca9db : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d0209f49b4483e86faa295dc024a2f : SUCCESS in 5m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/95d735a20d194060b65da9a4302a0ea4 : SUCCESS in 6m 06s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/044f653c3e5d40af97bc8dbf5f81cd20 : SUCCESS in 1h 19m 22s","accounts_in_message":[],"_revision_number":3},{"id":"91d158159c545014785bfba230688dd9df45efee","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-20 09:07:51.000000000","message":"Patch Set 3: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9b8e91481827660a303d07df591c4637fc99a901","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-27 21:33:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c024dfc2df657a38ae840c3b8e574d81f9b897ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 22:40:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b46970252f4ec88799d1b8a53ca75e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5867a274e347758572e82c457718d9 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d85d1d5c2ae14e9697bf5d62d541351d : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9078d9040664dcb97d5d4c81978b81f : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7040ecafa8684bd7bdce53644eca4eed : SUCCESS in 6m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dbf96273a6ab4419b70eb7840a35e0ed : SUCCESS in 6m 20s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b5e465b35ee042989a35c29556c5f19f : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":4},{"id":"39b9ebdc4141d92f61a8aee59d057b1336a97a9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 17:28:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8a327eab7a977ca23531672ceffd34dbeb1e5795","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 18:38:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc710d033a226113dc189a7198e6d04a641a7fbc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 18:39:17.000000000","message":"Abandoned\n\nI\u0027m going to push a different version that is easier to review","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fc2a2b1001761ee1ab36d876ac0034b7305caf37","revisions":{"8c6ebabe618e98ddb6c88b55bba00d2893218ebe":{"kind":"REWORK","_number":1,"created":"2026-01-15 22:54:10.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/64/973564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/64/973564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/1"}}},"commit":{"parents":[{"commit":"61cde4703cf032c7ce6bd8dce10c5822dbe12b09","subject":"Revert \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cde4703cf032c7ce6bd8dce10c5822dbe12b09"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-15 22:45:51.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-15 22:54:05.000000000","tz":-360},"subject":"Reapply \"Rework venv to support arbitrary schemas\"","message":"Reapply \"Rework venv to support arbitrary schemas\"\n\nThis reverts commit 61cde4703cf032c7ce6bd8dce10c5822dbe12b09 and\nfixes the issue that caused sporadic test failures. Previously,\nthe Backend._ovsdb_connection was being inherited by subclasses and\nwhen concurrency\u003d1, the multiple test classes would use the same\nconnection, but the venv would be torn down by other classes running\nin the same process when they finished, breaking the connectinon and\ntherefor the tests.\n\nFor reference, the text of the reapplied patches is\ncommit 7744650d02062b99115535fb02108b6eeab6f7b9\n\n    Improve Row stringification\n\n    Add the Row\u0027s uuid and support for handling converting Rows that\n    are part of sets and maps to UUIDs instead of logging the repr\n    which is the default Python pointer info.\n\n    python-ovs has a Row.__str__ and this is changing there as well.\n    Once the version that has it is our minimum required version,\n    row2str() can just return str(row).\n\ncommit 9c1790796d8094d386257ebd2909186a9d63352c\n\n    Rework venv to support arbitrary schemas\n\nChange-Id: Ic0c721fe088a53de98f1149e41ad5f72624be456\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8c6ebabe618e98ddb6c88b55bba00d2893218ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8c6ebabe618e98ddb6c88b55bba00d2893218ebe"}]},"branch":"refs/heads/master"},"896a9a121841419ce04fdd760605e6b2703ffacd":{"kind":"REWORK","_number":2,"created":"2026-01-19 22:51:18.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/64/973564/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/64/973564/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/2"}}},"commit":{"parents":[{"commit":"61cde4703cf032c7ce6bd8dce10c5822dbe12b09","subject":"Revert \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cde4703cf032c7ce6bd8dce10c5822dbe12b09"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-15 22:45:51.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-19 22:51:17.000000000","tz":0},"subject":"Reapply \"Rework venv to support arbitrary schemas\"","message":"Reapply \"Rework venv to support arbitrary schemas\"\n\nThis reverts commit 61cde4703cf032c7ce6bd8dce10c5822dbe12b09 and\nfixes the issue that caused sporadic test failures. Previously,\nthe Backend._ovsdb_connection was being inherited by subclasses and\nwhen concurrency\u003d1, the multiple test classes would use the same\nconnection, but the venv would be torn down by other classes running\nin the same process when they finished, breaking the connectinon and\ntherefor the tests.\n\nFor reference, the text of the reapplied patches is\ncommit 7744650d02062b99115535fb02108b6eeab6f7b9\n\n    Improve Row stringification\n\n    Add the Row\u0027s uuid and support for handling converting Rows that\n    are part of sets and maps to UUIDs instead of logging the repr\n    which is the default Python pointer info.\n\n    python-ovs has a Row.__str__ and this is changing there as well.\n    Once the version that has it is our minimum required version,\n    row2str() can just return str(row).\n\ncommit 9c1790796d8094d386257ebd2909186a9d63352c\n\n    Rework venv to support arbitrary schemas\n\nChange-Id: Ic0c721fe088a53de98f1149e41ad5f72624be456\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/896a9a121841419ce04fdd760605e6b2703ffacd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/896a9a121841419ce04fdd760605e6b2703ffacd"}]},"branch":"refs/heads/master"},"922521321492c47594c213cf1b8980346741a606":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-19 22:51:57.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/64/973564/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/64/973564/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/3"}}},"commit":{"parents":[{"commit":"61cde4703cf032c7ce6bd8dce10c5822dbe12b09","subject":"Revert \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cde4703cf032c7ce6bd8dce10c5822dbe12b09"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-15 22:45:51.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-19 22:51:57.000000000","tz":0},"subject":"Reapply \"Rework venv to support arbitrary schemas\"","message":"Reapply \"Rework venv to support arbitrary schemas\"\n\nThis reverts commit 61cde4703cf032c7ce6bd8dce10c5822dbe12b09 and\nfixes the issue that caused sporadic test failures. Previously,\nthe Backend._ovsdb_connection was being inherited by subclasses and\nwhen concurrency\u003d1, the multiple test classes would use the same\nconnection, but the venv would be torn down by other classes running\nin the same process when they finished, breaking the connectinon and\ntherefor the tests.\n\nFor reference, the text of the reapplied patches is\ncommit 7744650d02062b99115535fb02108b6eeab6f7b9\n\n    Improve Row stringification\n\n    Add the Row\u0027s uuid and support for handling converting Rows that\n    are part of sets and maps to UUIDs instead of logging the repr\n    which is the default Python pointer info.\n\n    python-ovs has a Row.__str__ and this is changing there as well.\n    Once the version that has it is our minimum required version,\n    row2str() can just return str(row).\n\ncommit 9c1790796d8094d386257ebd2909186a9d63352c\n\n    Rework venv to support arbitrary schemas\n\nCloses-Bug: #2132018\nChange-Id: Ic0c721fe088a53de98f1149e41ad5f72624be456\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/922521321492c47594c213cf1b8980346741a606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/922521321492c47594c213cf1b8980346741a606"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9be361824876af0702d0b578c8c56dc87764e867":{"kind":"REWORK","_number":4,"created":"2026-01-27 21:33:30.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/64/973564/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/64/973564/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/4"}}},"commit":{"parents":[{"commit":"b4b46458362e299682bbbd23016c29f8e5c0d761","subject":"windows: Add close() method to WaitQueue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b4b46458362e299682bbbd23016c29f8e5c0d761"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-07-10 17:49:08.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-27 21:25:04.000000000","tz":-360},"subject":"Reapply Rework venv to support arbitrary schemas","message":"Reapply Rework venv to support arbitrary schemas\n\nThis partially reverts 61cde4703cf032c7ce6bd8dce10c5822dbe12b09 and\nfixes the issue that caused sporadic test failures. Previously,\nthe Backend._ovsdb_connection was being inherited by subclasses and\nwhen concurrency\u003d1, the multiple test classes would use the same\nconnection, but the venv would be torn down by other classes running\nin the same process when they finished, breaking the connectinon and\ntherefor the tests.\n\nChange-Id: Ic0c721fe088a53de98f1149e41ad5f72624be456\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n(cherry picked from commit 9c1790796d8094d386257ebd2909186a9d63352c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9be361824876af0702d0b578c8c56dc87764e867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9be361824876af0702d0b578c8c56dc87764e867"}]},"branch":"refs/heads/master"},"fc2a2b1001761ee1ab36d876ac0034b7305caf37":{"kind":"REWORK","_number":5,"created":"2026-01-28 17:28:13.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/64/973564/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/64/973564/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/64/973564/5"}}},"commit":{"parents":[{"commit":"61cde4703cf032c7ce6bd8dce10c5822dbe12b09","subject":"Revert \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cde4703cf032c7ce6bd8dce10c5822dbe12b09"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-15 22:45:51.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-28 17:28:05.000000000","tz":-360},"subject":"Reapply \"Rework venv to support arbitrary schemas\"","message":"Reapply \"Rework venv to support arbitrary schemas\"\n\nThis reverts commit 61cde4703cf032c7ce6bd8dce10c5822dbe12b09 and\nfixes the issue that caused sporadic test failures. Previously,\nthe Backend._ovsdb_connection was being inherited by subclasses and\nwhen concurrency\u003d1, the multiple test classes would use the same\nconnection, but the venv would be torn down by other classes running\nin the same process when they finished, breaking the connectinon and\ntherefor the tests.\n\nFor reference, the text of the reapplied patches is\ncommit 7744650d02062b99115535fb02108b6eeab6f7b9\n\n    Improve Row stringification\n\n    Add the Row\u0027s uuid and support for handling converting Rows that\n    are part of sets and maps to UUIDs instead of logging the repr\n    which is the default Python pointer info.\n\n    python-ovs has a Row.__str__ and this is changing there as well.\n    Once the version that has it is our minimum required version,\n    row2str() can just return str(row).\n\ncommit 9c1790796d8094d386257ebd2909186a9d63352c\n\n    Rework venv to support arbitrary schemas\n\nChange-Id: Ic0c721fe088a53de98f1149e41ad5f72624be456\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fc2a2b1001761ee1ab36d876ac0034b7305caf37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fc2a2b1001761ee1ab36d876ac0034b7305caf37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
