)]}'
{"id":"openstack%2Fovsdbapp~974956","triplet_id":"openstack%2Fovsdbapp~master~I13fcdaf009e59d16c44e2af73c5cef89bf88f5e5","project":"openstack/ovsdbapp","branch":"master","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-02-09 22:11:21.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-02-09 21:12:19.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"I13fcdaf009e59d16c44e2af73c5cef89bf88f5e5","subject":"Re-apply Improve Row stringification","status":"MERGED","created":"2026-01-27 21:33:30.000000000","updated":"2026-02-09 22:12:14.000000000","submitted":"2026-02-09 22:11:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"974956","meta_rev_id":"e62f381491cefa3715c4df35a9aa66ab99a85058","_number":974956,"virtual_id_number":974956,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 22:11:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-02-06 09:48:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-09 21:12:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-09 21:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 22:33:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 10:38:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-02-05 12:54:00.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2026-02-06 09:48:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-02-09 21:12:19.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"90e917b72e2b6830762cafd0dcbe086dc608fecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-27 21:33:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60c932ae47d443a3a9eca58ab915ca2708995323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 22:33:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5938f59c90d0421c85392f2d49bfc7cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c0dc7221c5446ca8fe169cda0719d6c : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2365f2e3dbc2482e810ca075f7adc862 : SUCCESS in 2m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69371e2ab46a432683773e29ef2cb7ff : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2fb4a2da2b74b61995053b24f70e6bd : SUCCESS in 6m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3ce32791575d4566bd09473e3c8e6efa : SUCCESS in 6m 17s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/614d97faf7b4456cb989a64b7f8f4e47 : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":1},{"id":"19454008f8703a20f503af7481bca7756cc945c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 17:28:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8fae529d28c5187b753f5b57d2544562ae4e13bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 18:54:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a65451572e54dddbb7576b15b0b3223\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/90e0f46bf01844d4b72bcdcce56729ee : SUCCESS in 2m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d4edefd449854f1091133aa30aa59b94 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e834f99e5a4003b0aabb42ae953cc3 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/081bcf5a564144dbbb47008049f45f16 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0620301621254c788191d443431e3868 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0b2251500b241a696fbc1be4fc8d04b : SUCCESS in 8m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/433994a04d674e579fdd54d651ca8729 : SUCCESS in 5m 00s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a1142fee76b4441b826cdf47f068ca65 : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"12c74ec4bf7282d9e9f9ba93924e9aa7b68a79f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 19:25:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5852597bdefe56a63cd62a1a462492f9e24ecebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 20:33:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/738f1b19c2ed44ad9288d4043f5133bb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f8c2d795082745509ea8135d6ccb6f6e : SUCCESS in 2m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1e2db3a8ad8c4306b9355363393a387b : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/219a80bdf5364031b65e8f5e80da8950 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d166b0527084d10848663263e76af17 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07d77008177c4428a985e3d71ea10649 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d7c00ee7b6e4279b822311332bdbd62 : SUCCESS in 6m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d705dfa1c2ff4a7c9f8c9bbd2053b48f : SUCCESS in 4m 28s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d7912fd6060e448791c6c1847de196ce : SUCCESS in 1h 04m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9968412d4d08261e47435bea4424fdc70f8194d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 22:09:22.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"dbd28b37b56d510f6aafb447a6419c51b135fe21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 22:35:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"855b0f869a6e55922119c646ed84e43f4fd6f8e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 23:22:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/286a954e7cbe4858b0fc6f205128b710\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260abe1fa22743658f28ca8da71c5056 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b078e077f2e46a19b4e726181b63f18 : SUCCESS in 2m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07e9fd12305b47fc973f00359f832bcc : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c795b3b08c4960892b038ab8ebc6e6 : SUCCESS in 5m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/25217a684b8445e48cf88bf06e22bfbf : SUCCESS in 3m 38s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2e8da46e7bc0410890801117566100bc : SUCCESS in 46m 37s","accounts_in_message":[],"_revision_number":5},{"id":"fe32cd39864008c7b59e18aca5d4b0e444d9ccd2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-05 10:38:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e43c5eedbc429f9793df7bca8deb825b2e6f2ac9","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-05 12:54:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f102b1fb3880dda55ec78b19a44d78e26cb82811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-05 14:03:35.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d7ecb6c196be009df3182db74dd157819dfddcb7","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-05 14:22:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f27734615cd92b197e205cb84753f3e8bd3e066a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 15:42:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6375bf2b6f014a12a2e171815fc6b3fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd08b04c4b34911bee0d0ba301757f3 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23b4d4abeffd4d108927c6f7d257bcc1 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ef6794e8b454df184651effc0db567d : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc7f49c360443aeab7a32f87da46911 : SUCCESS in 6m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04a80321177d454da59c5eee89f1bab6 : SUCCESS in 5m 44s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/92c12fa0fc99477c8f88f4f24912b776 : SUCCESS in 1h 22m 21s","accounts_in_message":[],"_revision_number":6},{"id":"ef3dfbf8fef937dd18b4d1e712bfddf900de16eb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-06 09:48:13.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1732298e29b08a2160a9073edeb68fe78f13af90","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-09 21:12:19.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"01a88095a36dde902ae423f41632072239431186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 21:12:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"01da79ce7e262d857dceba05e0acee88241a76f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 22:11:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24a00479d9242d58dd5c10919a7cedf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f9cc57e19e48c9a8aa44cefc4c4d8f : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecbb2b87afaf4bc482c2680c06087b6c : SUCCESS in 1m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b4c408e16894922b8173dafa549d966 : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc1b7b3379241f59dec823a9b4a6128 : SUCCESS in 6m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/310eb1b22d6b4c7a953ceb1c4cd41e47 : SUCCESS in 3m 38s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f4c2a57e2cb0483fb05f9e0d6cab53a0 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":6},{"id":"fc26f24a5508314255222ee3903c8d7a36248d96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 22:11:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"e62f381491cefa3715c4df35a9aa66ab99a85058","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 22:12:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/422da1ee5d874eb9a5bf474887fe7784\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d097bf5f72e4cebae4ea939fa63da9c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"fdc549365c7c21bd98ebb0e49db9f4f81fa82035","revisions":{"43fe3b2e21e1633bd5e270fd12920b6ec6baf5ae":{"kind":"REWORK","_number":1,"created":"2026-01-27 21:33:30.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/56/974956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/56/974956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/1"}}},"commit":{"parents":[{"commit":"9be361824876af0702d0b578c8c56dc87764e867","subject":"Reapply Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9be361824876af0702d0b578c8c56dc87764e867"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-27 21:33:15.000000000","tz":-360},"subject":"Re-apply Improve Row stringification","message":"Re-apply Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nThis reapplies the patch that was reverted as part of\n61cde4703cf032c7ce6bd8dce10c5822dbe12b09, after fixing the\nits parent patch.\n\nChange-Id: I13fcdaf009e59d16c44e2af73c5cef89bf88f5e5\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n(cherry picked from commit 7744650d02062b99115535fb02108b6eeab6f7b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/43fe3b2e21e1633bd5e270fd12920b6ec6baf5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/43fe3b2e21e1633bd5e270fd12920b6ec6baf5ae"}]},"branch":"refs/heads/master"},"99c566f22b4c935341e5fa2372cce22ba41e1b44":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-28 17:28:24.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/56/974956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/56/974956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/2"}}},"commit":{"parents":[{"commit":"fc2a2b1001761ee1ab36d876ac0034b7305caf37","subject":"Reapply \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fc2a2b1001761ee1ab36d876ac0034b7305caf37"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-28 17:28:24.000000000","tz":0},"subject":"Re-apply Improve Row stringification","message":"Re-apply Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nThis reapplies the patch that was reverted as part of\n61cde4703cf032c7ce6bd8dce10c5822dbe12b09, after fixing the\nits parent patch.\n\nChange-Id: I13fcdaf009e59d16c44e2af73c5cef89bf88f5e5\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n(cherry picked from commit 7744650d02062b99115535fb02108b6eeab6f7b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/99c566f22b4c935341e5fa2372cce22ba41e1b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/99c566f22b4c935341e5fa2372cce22ba41e1b44"}]},"branch":"refs/heads/master","description":"Rebase"},"e6ac9b8fb23ac335a835af7a58a98de2bd4c2142":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-28 19:25:40.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/56/974956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/56/974956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/3"}}},"commit":{"parents":[{"commit":"043f752fa24264f8aa0c1ac00f4f17ffc7d443d8","subject":"Re-apply Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/043f752fa24264f8aa0c1ac00f4f17ffc7d443d8"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-28 19:25:40.000000000","tz":0},"subject":"Re-apply Improve Row stringification","message":"Re-apply Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nThis reapplies the patch that was reverted as part of\n61cde4703cf032c7ce6bd8dce10c5822dbe12b09, after fixing the\nits parent patch.\n\nChange-Id: I13fcdaf009e59d16c44e2af73c5cef89bf88f5e5\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n(cherry picked from commit 7744650d02062b99115535fb02108b6eeab6f7b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e6ac9b8fb23ac335a835af7a58a98de2bd4c2142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e6ac9b8fb23ac335a835af7a58a98de2bd4c2142"}]},"branch":"refs/heads/master","description":"Rebase"},"588e976bee7aac350f398bab5f2a8599b09a25a4":{"kind":"NO_CHANGE","_number":4,"created":"2026-01-28 22:09:22.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/56/974956/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/56/974956/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/4"}}},"commit":{"parents":[{"commit":"f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0","subject":"Re-apply Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-28 22:09:22.000000000","tz":0},"subject":"Re-apply Improve Row stringification","message":"Re-apply Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nThis reapplies the patch that was reverted as part of\n61cde4703cf032c7ce6bd8dce10c5822dbe12b09, after fixing the\nits parent patch.\n\nChange-Id: I13fcdaf009e59d16c44e2af73c5cef89bf88f5e5\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n(cherry picked from commit 7744650d02062b99115535fb02108b6eeab6f7b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/588e976bee7aac350f398bab5f2a8599b09a25a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/588e976bee7aac350f398bab5f2a8599b09a25a4"}]},"branch":"refs/heads/master","description":"Rebase"},"0f2d22748c8159c32e8c7a60d091b8ea03c19b3d":{"kind":"REWORK","_number":5,"created":"2026-01-28 22:35:31.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/56/974956/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/56/974956/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/5"}}},"commit":{"parents":[{"commit":"f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0","subject":"Re-apply Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-28 22:33:03.000000000","tz":-360},"subject":"Re-apply Improve Row stringification","message":"Re-apply Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nThis reapplies the patch that was reverted as part of\n61cde4703cf032c7ce6bd8dce10c5822dbe12b09, after fixing the\nits parent patch.\n\nChange-Id: I13fcdaf009e59d16c44e2af73c5cef89bf88f5e5\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n(cherry picked from commit 7744650d02062b99115535fb02108b6eeab6f7b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0f2d22748c8159c32e8c7a60d091b8ea03c19b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0f2d22748c8159c32e8c7a60d091b8ea03c19b3d"}]},"branch":"refs/heads/master"},"fdc549365c7c21bd98ebb0e49db9f4f81fa82035":{"kind":"REWORK","_number":6,"created":"2026-02-05 14:03:35.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/56/974956/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/56/974956/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/56/974956/6"}}},"commit":{"parents":[{"commit":"f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0","subject":"Re-apply Rework venv to support arbitrary schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-06-18 21:56:40.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-02-05 14:03:35.000000000","tz":0},"subject":"Re-apply Improve Row stringification","message":"Re-apply Improve Row stringification\n\nAdd the Row\u0027s uuid and support for handling converting Rows that\nare part of sets and maps to UUIDs instead of logging the repr\nwhich is the default Python pointer info.\n\npython-ovs has a Row.__str__ and this is changing there as well.\nOnce the version that has it is our minimum required version,\nrow2str() can just return str(row).\n\nThis reapplies the patch that was reverted as part of\n61cde4703cf032c7ce6bd8dce10c5822dbe12b09, after fixing the\nits parent patch.\n\nChange-Id: I13fcdaf009e59d16c44e2af73c5cef89bf88f5e5\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n(cherry picked from commit 7744650d02062b99115535fb02108b6eeab6f7b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fdc549365c7c21bd98ebb0e49db9f4f81fa82035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fdc549365c7c21bd98ebb0e49db9f4f81fa82035"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
