)]}'
{"id":"openstack%2Fovsdbapp~975087","triplet_id":"openstack%2Fovsdbapp~master~I38033de268af2e53daceeec9d9005be52a317cfb","project":"openstack/ovsdbapp","branch":"master","topic":"resolve_venv_conflicts","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-02-04 19:29:57.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-02-04 19:58:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I38033de268af2e53daceeec9d9005be52a317cfb","subject":"Re-apply Rework venv to support arbitrary schemas","status":"MERGED","created":"2026-01-28 18:49:25.000000000","updated":"2026-02-04 19:59:50.000000000","submitted":"2026-02-04 19:58:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"975087-resolve_venv_conflicts","meta_rev_id":"f34054cbf6e423a148a61cc7abcb4d1535477abc","_number":975087,"virtual_id_number":975087,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-04 19:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-02-04 18:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2026-01-29 18:43:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-02-04 18:29:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-28 21:03:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 23:40:46.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2026-02-04 18:29:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"e9a617c316543f487bc71ac7ac46933ff66eddef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 18:49:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c045ca0f1418660f276a6a3e335078a99976fa21","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 18:50:04.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"008b696f2f3039fcdf8fc608caa1230f6e36fac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 19:23:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6bc47ad10f473aa96d71e2ae9a9d58fdff87e196","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 19:25:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"960355742cd5d380a3479cfb437b0062a8857a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 21:03:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679f7caceb154189a15930ba3f479242\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f5cf69c6b5043db8c4f2b61d19d5129 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bca996fee4148598f7858617dd86b97 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ac124426b634e9c9aeabf740e799428 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41df11dd9aaa470cb31909c63d21b9f4 : SUCCESS in 6m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9708848cff494e3c8450983b347d4417 : SUCCESS in 6m 04s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d1105ab0cfae417ca26a1ac92eaddb5b : SUCCESS in 1h 25m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b19082e47befdaf6f3060f3717772ef1bb9e99c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-28 22:09:10.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9e20be8b8a5e07ffe0f0dc8254b9bfaeb477ea35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 23:32:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8da6e4e879e244bf84825982522bcab3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add2ff24774f42f9880f25053409356b : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12f41f5d1f404b5eb4fc69d83cee80d1 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6aa721275e0b4c129d9bf0dc2f31924e : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db96ca6bdf5b46ebacea6c230fae3167 : SUCCESS in 3m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3f85a39da1094882b9d6d1dbb3a33dac : SUCCESS in 5m 42s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ae6b2de1f60f43478f363e93bc1844b7 : SUCCESS in 1h 22m 12s","accounts_in_message":[],"_revision_number":3},{"id":"6d49857aa113c003e73de1c318bfc008a069c4c2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-01-28 23:40:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3d08029d7ca88229a7e6b10c3461cd1cf592dfc1","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-01-29 18:43:44.000000000","message":"Patch Set 3: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e5e081816332501d90664d58440220a148b143bb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-29 19:07:35.000000000","message":"Patch Set 3:\n\n(12 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4dcd1a0198cc3d4a90f17cab4c9b8a5b2d8e3073","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-01-30 14:30:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6a8c1c53eeff7c44d292fb64ce36f6be53cb7ffb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-04 18:29:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"99b131cc0d1f37e5a4a15e5d60ccc62b2a200bca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 18:29:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d5e9ddfcb87dcdc0d6377affa81e2701ddfdd793","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-02-04 19:29:57.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4ccac34ba48582b9bd4148002ed44375af10a418","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 19:58:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d1229892ab64b289b88be05b9ff1381\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac3c27e858c4b948e3eab117c5eed09 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8fb859963e143e98cd7d39611ed6f53 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76bb3a61213c49efbde02bcbbc2b13a3 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69d3a32c88004b33a529286728514027 : SUCCESS in 6m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d4afea4d54a941aca392a1ad2c1e9410 : SUCCESS in 6m 01s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b0013c7c3250451b88788e71ddb6cb2c : SUCCESS in 1h 15m 41s","accounts_in_message":[],"_revision_number":3},{"id":"bfdc579f552e3f574de1e1b411e38bdfb62fe958","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 19:58:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f34054cbf6e423a148a61cc7abcb4d1535477abc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 19:59:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24069bd9a2e641dc8d466e0f1180e312\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e44c5480e9c4557ad97dfbb5e7172a5 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0","revisions":{"5c3d644672f858ca6d887f5ba984d5a4ad5a6bf1":{"kind":"REWORK","_number":1,"created":"2026-01-28 18:49:25.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/87/975087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/87/975087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/1"}}},"commit":{"parents":[{"commit":"61cde4703cf032c7ce6bd8dce10c5822dbe12b09","subject":"Revert \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cde4703cf032c7ce6bd8dce10c5822dbe12b09"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-07-10 17:49:08.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-28 18:48:50.000000000","tz":-360},"subject":"Re-apply Rework venv to support arbitrary schemas","message":"Re-apply Rework venv to support arbitrary schemas\n\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\nChange-Id: I38033de268af2e53daceeec9d9005be52a317cfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5c3d644672f858ca6d887f5ba984d5a4ad5a6bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5c3d644672f858ca6d887f5ba984d5a4ad5a6bf1"}]},"branch":"refs/heads/master"},"043f752fa24264f8aa0c1ac00f4f17ffc7d443d8":{"kind":"REWORK","_number":2,"created":"2026-01-28 19:23:02.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/87/975087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/87/975087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/2"}}},"commit":{"parents":[{"commit":"61cde4703cf032c7ce6bd8dce10c5822dbe12b09","subject":"Revert \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cde4703cf032c7ce6bd8dce10c5822dbe12b09"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-07-10 17:49:08.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-28 19:22:51.000000000","tz":-360},"subject":"Re-apply Rework venv to support arbitrary schemas","message":"Re-apply Rework venv to support arbitrary schemas\n\nThis partially reverts 61cde4703cf032c7ce6bd8dce10c5822dbe12b09 and\nfixes the issue that caused sporadic test failures. Previously,\nthe Backend._ovsdb_connection was being inherited by subclasses and\nwhen concurrency\u003d1, the multiple test classes would use the same\nconnection, but the venv would be torn down by other classes running\nin the same process when they finished, breaking the connectinon and\ntherefor the tests.\n\nThis also includes Jakub Libosvar\u0027s fix for handling an empty search\npath in schema_path.\n\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\nChange-Id: I38033de268af2e53daceeec9d9005be52a317cfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/043f752fa24264f8aa0c1ac00f4f17ffc7d443d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/043f752fa24264f8aa0c1ac00f4f17ffc7d443d8"}]},"branch":"refs/heads/master"},"f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-28 22:09:10.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/87/975087/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/87/975087/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/87/975087/3"}}},"commit":{"parents":[{"commit":"61cde4703cf032c7ce6bd8dce10c5822dbe12b09","subject":"Revert \"Rework venv to support arbitrary schemas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/61cde4703cf032c7ce6bd8dce10c5822dbe12b09"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-07-10 17:49:08.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-01-28 22:09:10.000000000","tz":0},"subject":"Re-apply Rework venv to support arbitrary schemas","message":"Re-apply Rework venv to support arbitrary schemas\n\nThis partially reverts 61cde4703cf032c7ce6bd8dce10c5822dbe12b09 and\nfixes the issue that caused sporadic test failures. Previously,\nthe Backend._ovsdb_connection was being inherited by subclasses and\nwhen concurrency\u003d1, the multiple test classes would use the same\nconnection, but the venv would be torn down by other classes running\nin the same process when they finished, breaking the connectinon and\ntherefor the tests.\n\nThis also includes Jakub Libosvar\u0027s fix for handling an empty search\npath in schema_path.\n\nCloses-Bug: #2132018\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\nChange-Id: I38033de268af2e53daceeec9d9005be52a317cfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f34e599aedd1dcc2bf014b8bac95f0f5c6ce83e0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
