)]}'
{"id":"openstack%2Fovsdbapp~979081","triplet_id":"openstack%2Fovsdbapp~master~I016e2fd5c27ee0748c5e4e48251443e634fd5c91","project":"openstack/ovsdbapp","branch":"master","topic":"ovn-func-bump","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-03-06 15:10:24.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-03-06 17:17:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I016e2fd5c27ee0748c5e4e48251443e634fd5c91","subject":"Set up column when simulating port binding","status":"MERGED","created":"2026-03-05 20:57:24.000000000","updated":"2026-03-06 17:18:05.000000000","submitted":"2026-03-06 17:17:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979081-ovn-func-bump","meta_rev_id":"6d0e454a2507b17d7070dfa3db19a1e2a4b2146d","_number":979081,"virtual_id_number":979081,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-06 17:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-06 14:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-03-06 15:10:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-06 15:05:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-06 16:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-03-06 14:47:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-05 21:41:44.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-03-05 22:54:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-06 14:47:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-03-06 15:05:17.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"}],"messages":[{"id":"c2003ce99501c752e367f33af602734459d6a288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-05 20:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61e11f70f8af4ec8c2144c0b735273df5ada97b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-05 21:23:22.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9784f1987d7a60206708ed409de8a0abf8437158","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-05 21:41:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"031ed94235fb98eec21488caf8bf609cba3f9cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 22:54:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29f1ed3884a743df801cc4f138a0b0c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc34fcaafadc43479225b904f522509e : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bcf6fd85c064b6585c3b26e2a63ec6f : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa036d05f5194edb8c7c69b3976d0642 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8a5617ec16ea4211b95eabebb74745f8 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22da70b716274dc9b390de11479d6af1 : SUCCESS in 6m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9f88f7d890504de784dfbceca6b8b5f6 : FAILURE in 6m 10s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/34d94dfcf44f40c7bfaa595eb695e2ad : SUCCESS in 1h 25m 46s","accounts_in_message":[],"_revision_number":2},{"id":"261d9f4998285824f7a631559d89a7c342e46d04","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 00:02:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5c5eb7e19a279ab6b91a0d36e6a345e873c1ff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 00:03:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fadbfa48a2d2d873f7489c586527619a933d3790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 01:31:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c160b044b054a0ca4e27329c61e00db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82b5fe34619d4f58b6d0f0cef209f4ab : FAILURE in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/274f658bf0954157a146b04dce2efd93 : FAILURE in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a17b1c80e3ad4705a7e16a1b36177d3d : FAILURE in 2m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c0c95d42897245388c05b4a27c4b735f : FAILURE in 3m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e9ba9821324c458c2670077bfc2d22 : FAILURE in 3m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a954ab6c8a8042e393e39f87ca6e6fa6 : FAILURE in 1m 59s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aa8fa87f3eb74bb7b07519c890537426 : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2834efc2754cf043cdfd3e424a17fc8d18349475","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-06 14:11:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9d69d5cac87594c184c1c4c507a83535acb28ea9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-06 14:47:51.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"ce9dd30b4654615376359e32f3d0db1a39560cee","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2026-03-06 15:05:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ab3592d04af0ffb59c4de0804e3c33990298f5b4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-03-06 15:10:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65f3eb47ffac239a153c714256548669f6d46093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 15:43:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f99872b26b54987b542e3cec522252d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ef495273fd4dbf9907ecf6980f19b5 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f766ec7ed84d41e1a321089910a60ac0 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29c773ffc4e3438892047773020d0c7b : SUCCESS in 2m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/67704e6df89849299cd554d48ec5739e : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de0f74c7b84f40c086037b998ef8c8cf : SUCCESS in 9m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4ce6809363c54c7f8026581ab979c03d : SUCCESS in 6m 26s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c7b8b7862d6e4bc0b8315434c1a34ee4 : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":3},{"id":"04ee6c4671646c678c17adaa226a71b2eec957bf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-06 16:07:09.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"62dfeade7c1e0d20f726a88d8fbb5e045e92de71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 16:07:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"77d4a2cd8a20d2e4eec67eb3cc1e96b077902bb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 17:17:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5019e698f2f641fc9866318e7d7f9020\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df673168c5543ab96d1a3756b232a20 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9da41da5f0494ab394f50f27e9438b96 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f96cd0937a6148fdb3df415d148ad5ef : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a6d9dd683d4823bb5ac71da1b0edd2 : SUCCESS in 6m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c6a74f2a709649979f2e5326c3c8a964 : SUCCESS in 9m 30s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6894b5f3ac2d45d0a22feee456a82bf7 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":3},{"id":"7956096bb861923f9ef0150e704bb2eb72c89f4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 17:17:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6d0e454a2507b17d7070dfa3db19a1e2a4b2146d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 17:18:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2134eb57ac304fc8a89d60307dccd08e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4fe8bcb275e4e1aae847e335deed4dc : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a8dcb36c33fe7720e20de4d85bad2f461d102a01","revisions":{"687d6d29b6269d46e0984ea62733585e284211e7":{"kind":"REWORK","_number":1,"created":"2026-03-05 20:57:24.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/81/979081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/81/979081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/1"}}},"commit":{"parents":[{"commit":"9df3d19c236179a5323a70c21544b111e16a31bc","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9df3d19c236179a5323a70c21544b111e16a31bc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-05 20:38:19.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-05 20:57:17.000000000","tz":-300},"subject":"Set up column when simulating port binding","message":"Set up column when simulating port binding\n\nThe current northd in OVN 26.03 no longer considers port-up-notif\nfeature configured on the chassis [1] and in order to set an LSP.up column\nit requires the corresponding PB.chassis AND PB.up being set, as opposed\nto previous versions that considered only PB.chassis, if port-up-notif\nwas not set.\n\nThe LspBindCommand is used as an ovn-controller binding simulation and\nhence since 24.03, where the port-notif-up was deprecated, should also\nset the PB.up column along with the chassis.\n\n[1] https://github.com/ovn-org/ovn/commit/83cf9cc9b35cb2a98706b9df37a6d20b31287255\n\nChange-Id: I016e2fd5c27ee0748c5e4e48251443e634fd5c91\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/687d6d29b6269d46e0984ea62733585e284211e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/687d6d29b6269d46e0984ea62733585e284211e7"}]},"branch":"refs/heads/master"},"2082b552c08627ea0db32422d422f1ef578bc01c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-05 21:23:22.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/81/979081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/81/979081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/2"}}},"commit":{"parents":[{"commit":"9df3d19c236179a5323a70c21544b111e16a31bc","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9df3d19c236179a5323a70c21544b111e16a31bc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-05 20:38:19.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-05 21:23:22.000000000","tz":0},"subject":"Set up column when simulating port binding","message":"Set up column when simulating port binding\n\nThe current northd in OVN 26.03 no longer considers port-up-notif\nfeature configured on the chassis [1] and in order to set an LSP.up\ncolumn it requires the corresponding PB.chassis AND PB.up being set,\nas opposed to previous versions that considered only PB.chassis, if\nport-up-notif was not set.\n\nThe LspBindCommand is used as an ovn-controller binding simulation and\nhence since 24.03, where the port-notif-up was deprecated, should also\nset the PB.up column along with the chassis.\n\nThis also mirrors what ovn-sbctl lsp-bind does:\nhttps://github.com/ovn-org/ovn/blob/aab84f86503050064c970fc7f62764598e1b9c67/utilities/ovn-sbctl.c#L549\n\n[1] https://github.com/ovn-org/ovn/commit/83cf9cc9b35cb2a98706b9df37a6d20b31287255\n\nChange-Id: I016e2fd5c27ee0748c5e4e48251443e634fd5c91\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2082b552c08627ea0db32422d422f1ef578bc01c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2082b552c08627ea0db32422d422f1ef578bc01c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a8dcb36c33fe7720e20de4d85bad2f461d102a01":{"kind":"REWORK","_number":3,"created":"2026-03-06 00:03:53.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/81/979081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/81/979081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/81/979081/3"}}},"commit":{"parents":[{"commit":"9df3d19c236179a5323a70c21544b111e16a31bc","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9df3d19c236179a5323a70c21544b111e16a31bc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-05 20:38:19.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-06 00:01:43.000000000","tz":0},"subject":"Set up column when simulating port binding","message":"Set up column when simulating port binding\n\nThe current northd in OVN 26.03 no longer considers port-up-notif\nfeature configured on the chassis [1] and in order to set an LSP.up\ncolumn it requires the corresponding PB.chassis AND PB.up being set,\nas opposed to previous versions that considered only PB.chassis, if\nport-up-notif was not set.\n\nThe LspBindCommand is used as an ovn-controller binding simulation and\nhence since 24.03, where the port-notif-up was deprecated, should also\nset the PB.up column along with the chassis.\n\nThis also mirrors what ovn-sbctl lsp-bind does:\nhttps://github.com/ovn-org/ovn/blob/aab84f86503050064c970fc7f62764598e1b9c67/utilities/ovn-sbctl.c#L549\n\n[1] https://github.com/ovn-org/ovn/commit/83cf9cc9b35cb2a98706b9df37a6d20b31287255\n\nChange-Id: I016e2fd5c27ee0748c5e4e48251443e634fd5c91\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a8dcb36c33fe7720e20de4d85bad2f461d102a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a8dcb36c33fe7720e20de4d85bad2f461d102a01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
