)]}'
{"id":"openstack%2Fovsdbapp~990589","triplet_id":"openstack%2Fovsdbapp~master~I3b62d7b305bcfb1273ebec3d13b2b5f1aef08ed5","project":"openstack/ovsdbapp","branch":"master","topic":"asyncio","attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-05-29 13:36:08.000000000","reason":"Vote got outdated and was removed: Workflow-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I3b62d7b305bcfb1273ebec3d13b2b5f1aef08ed5","subject":"WIP Implement asyncio Connections","status":"NEW","created":"2026-05-29 04:24:02.000000000","updated":"2026-05-29 14:27:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"abbac327e92d2334971f9e995178c45f3c19f935","_number":990589,"virtual_id_number":990589,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-29 14:27:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-29 05:55:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0403a50fa4c585deea7ec33445f17f68b06a53b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-29 04:24:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"511bb1826ec23a91a761f7473a4883d8e00e758f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-29 04:24:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3561f6353ff233e6372482e7d44c15135d022af4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 05:55:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b168cda2ec74b2483f430efd52a940b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac47bc9b9969447a8cac425e2e0595f6 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8faab9ad43a249d69a3ec8a37ef7c349 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a09c0fbb9de4340be1a566fcf3cd985 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/046fcffa23d44a26bba5c5578a5f60a9 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66f285cc71c43f9bf4f5235e3caa3a3 : FAILURE in 5m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8750aa52778445f0b2de4ad8e17bc172 : SUCCESS in 5m 12s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f31bc1b64622481abf1d4ad4ac84be55 : SUCCESS in 1h 26m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3a112ed0270a427a90ce175c05a260a5 : SUCCESS in 21m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b1b27736711bc74888bba2973a5373a44310eeaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-05-29 13:36:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"abbac327e92d2334971f9e995178c45f3c19f935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:27:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aa4e41478234156ac4a1cbf4b802472\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67acf8527bb641bcaeeeafe6d0a1333f : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eee446dc59a14cb18df039d85c72a31f : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b69124412d634442967c74c02cc0b6e2 : SUCCESS in 2m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/44dd28e432c343d8967d9a8f8dc41d50 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6c7555ced9446fa9fe0f834e396de0 : SUCCESS in 4m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b49912213d3b45b4be76b8a9890458dd : SUCCESS in 6m 16s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b3a261ac6fd7416cae58a2075f9cf47f : SUCCESS in 48m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ec1f05d50f364a26a5624103cab84624 : SUCCESS in 40m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f62e177380dbd150e1ed9f83b76b46d70489bd13","revisions":{"4f0b49a8bd5474921985cd1ce389e385d2cb367b":{"kind":"REWORK","_number":1,"created":"2026-05-29 04:24:02.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/89/990589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/89/990589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/990589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/990589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/990589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/89/990589/1"}}},"commit":{"parents":[{"commit":"e5fe03382e59c8b1df0442f4432d1b2bf22bb53d","subject":"Add Neutron functional test to the gate queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e5fe03382e59c8b1df0442f4432d1b2bf22bb53d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-05-28 20:33:41.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-05-29 04:23:38.000000000","tz":-300},"subject":"WIP Implement asyncio Connections","message":"WIP Implement asyncio Connections\n\nThis is currently just a backup checkin.\n\nAssisted-By: Claude Sonnet 4.6 (1M context) \u003cnoreply@anthropic.com\u003e\nChange-Id: I3b62d7b305bcfb1273ebec3d13b2b5f1aef08ed5\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4f0b49a8bd5474921985cd1ce389e385d2cb367b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4f0b49a8bd5474921985cd1ce389e385d2cb367b"}]},"branch":"refs/heads/master"},"f62e177380dbd150e1ed9f83b76b46d70489bd13":{"kind":"REWORK","_number":2,"created":"2026-05-29 13:36:08.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/89/990589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/89/990589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/990589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/990589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/89/990589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/89/990589/2"}}},"commit":{"parents":[{"commit":"e5fe03382e59c8b1df0442f4432d1b2bf22bb53d","subject":"Add Neutron functional test to the gate queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e5fe03382e59c8b1df0442f4432d1b2bf22bb53d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-05-28 20:33:41.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2026-05-29 13:30:40.000000000","tz":-300},"subject":"WIP Implement asyncio Connections","message":"WIP Implement asyncio Connections\n\nThis is currently just a backup checkin.\n\nAssisted-By: Claude Sonnet 4.6 (1M context) \u003cnoreply@anthropic.com\u003e\nChange-Id: I3b62d7b305bcfb1273ebec3d13b2b5f1aef08ed5\nSigned-off-by: Terry Wilson \u003ctwilson@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f62e177380dbd150e1ed9f83b76b46d70489bd13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f62e177380dbd150e1ed9f83b76b46d70489bd13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
