)]}'
{"id":"openstack%2Fovsdbapp~994611","triplet_id":"openstack%2Fovsdbapp~master~Ic8c098d22f3cfc3fadd7f7e1699c1f9a7ea57c9f","project":"openstack/ovsdbapp","branch":"master","topic":"fix/filter-matching-row-events","attention_set":{},"removed_from_attention_set":{"39125":{"account":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"last_update":"2026-06-26 09:16:02.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-06-24 19:31:24.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}},"hashtags":[],"change_id":"Ic8c098d22f3cfc3fadd7f7e1699c1f9a7ea57c9f","subject":"Don\u0027t call frozen_row if there are no matching events","status":"MERGED","created":"2026-06-24 09:37:58.000000000","updated":"2026-06-26 09:16:58.000000000","submitted":"2026-06-26 09:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994611-fix/filter-matching-row-events","meta_rev_id":"f8d465e09af64de465e0fcbf54720d82ed86f760","_number":994611,"virtual_id_number":994611,"owner":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-26 09:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-06-26 08:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-24 19:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2026-06-26 08:27:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 11:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 13:48:20.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2026-06-24 19:31:24.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-06-26 08:27:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"e7356c870337e16a7a1f71eae9a95d034439b7f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"date":"2026-06-24 09:37:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f63e9b9f4845919b8715775f78daca896a7ec7df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 11:07:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27330ea41c0c48e48deb1e9a90107298\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41764d2eecdf4dcbb25a4e9233038c27 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ba57fbde6514ad689520cec8c8b7b39 : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c1b639526ea4e73b4dcf95cf0f931b4 : SUCCESS in 2m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/770c983f7de74318851bc8072201a3e8 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7a239d0e7bd4371b939eb24b1a6efa3 : SUCCESS in 7m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9a3d32c90a524ce8aa1fe886f8c91668 : SUCCESS in 6m 52s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9cc7153add614ec28bd895317bc5151b : SUCCESS in 1h 23m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d47df74e64b847c587d688225c9c95c8 : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":1},{"id":"41d4c3b82c8629199f603c394d4f688c6b5aafb9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-24 13:48:20.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3a474361d3d5e8ad27b1f91398694cff899bcc59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"date":"2026-06-24 17:44:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3948f2fcd65dfc691acb687b42e05f52344e9a14","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"date":"2026-06-24 17:49:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6aa268db2132b646ea356ac18566a7340ad93c8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 19:13:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80d065eeae87414dbe306f6392274487\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66d3637c4e954ff78c0064d3749a9422 : SUCCESS in 2m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ed88728b6944475849200b286924e14 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdee6e6a70a249b7ac771a9160997fa3 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e1b89c7793f0434daf6169f8b2b685c8 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1adfec986d35403e85047c71d493cf65 : SUCCESS in 10m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a8405d1918d047b39c6bdf82b8ad4f13 : SUCCESS in 6m 54s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4f7045bf6d30472697f2d48871ea9966 : SUCCESS in 1h 22m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/22da187630ed4b85908802076b709367 : SUCCESS in 23m 21s","accounts_in_message":[],"_revision_number":2},{"id":"168d96d9d6cb2d75804bcce1af922ef4625046be","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-06-24 19:31:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"56d613fdf516eac22501e0010dee65233cd327e6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-26 08:27:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"19f1a7f44e4182ecfe2bf11796b4e5d4181bb225","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-26 08:27:44.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"63dff35009c0a68e8813c87911f036a601018faa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 08:28:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a3c8b6058fcf56eacbb19c41cf13424f4fa4b36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 09:16:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f750fefac2304646a5d0e73ae76febb3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e489c791bb634ec7b8368ff8e67a73bf : SUCCESS in 1m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21de47a30f5542efaee12006322943fe : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb3cb22fe94f474c8e4935e326c15924 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42253b60d0714a76bad7743669b7c126 : SUCCESS in 8m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/93d6a6dc48974b698938b3d82ef5a92a : SUCCESS in 7m 37s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8798c1bd3f574201b036e3fd4d7bc85e : SUCCESS in 47m 09s","accounts_in_message":[],"_revision_number":2},{"id":"734e096a91175f0000fc3e7fa5b7d7f4cf1bb52a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 09:16:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f8d465e09af64de465e0fcbf54720d82ed86f760","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 09:16:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03ccfc051ae34fcba359e73b93d162b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa181a1c4cb04896889ad8ab63bdcec7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b51e6cb04122541598af4e2cad766ac05ae32b46","revisions":{"1edaaea32b8132576f3ab91ed6d7b535ea0175d6":{"kind":"REWORK","_number":1,"created":"2026-06-24 09:37:58.000000000","uploader":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"ref":"refs/changes/11/994611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/11/994611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/11/994611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/11/994611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/11/994611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/11/994611/1"}}},"commit":{"parents":[{"commit":"10c1c984a138a78f29b468ad50bbd365fd247c2d","subject":"Merge \"Add asdict() method to RowView\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/10c1c984a138a78f29b468ad50bbd365fd247c2d"}]}],"author":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-06-23 12:52:23.000000000","tz":120},"committer":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-06-24 09:19:01.000000000","tz":120},"subject":"Don\u0027t call frozen_row if there are no matching events","message":"Don\u0027t call frozen_row if there are no matching events\n\nThe ovs_idl.RowEventHandler currently calls frozen_row to snapshot rows\nduring event handling even if there is no matching event.\nThis is not a problem in most cases as neutron already wraps this with\nits OvnDbNotifyHandler which similarly does the matching before calling\nfrozen_row. Additionally the hash ring mechanism on the neutron side\nspreads events across multiple workers.\n\nThe neutron maintenance worker does however use the RowEventHandler\ndirectly for its northbound and southbound IDL and also does not\nparticipate in the hash ring. This causes it to process all events and\nclone all rows from update notifications. In larger deployments this\ncauses the maintenance task to spend the majority of its time inside\nfrozen_row. This in turn can cause issues with disconnects due to\ninactivity probe timeouts and transactions timing out on the neutron\nside as the maintenance worker struggles to keep up with event\nprocessing.\n\nAdd the necessary filtering right here on the RowEventHandler instead\nof patching users on the neutron side individually.\n\nChange-Id: Ic8c098d22f3cfc3fadd7f7e1699c1f9a7ea57c9f\nSigned-off-by: Felix Moebius \u003cfelix.moebius@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1edaaea32b8132576f3ab91ed6d7b535ea0175d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1edaaea32b8132576f3ab91ed6d7b535ea0175d6"}]},"branch":"refs/heads/master"},"b51e6cb04122541598af4e2cad766ac05ae32b46":{"kind":"REWORK","_number":2,"created":"2026-06-24 17:44:44.000000000","uploader":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"ref":"refs/changes/11/994611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/11/994611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/11/994611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/11/994611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/11/994611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/11/994611/2"}}},"commit":{"parents":[{"commit":"10c1c984a138a78f29b468ad50bbd365fd247c2d","subject":"Merge \"Add asdict() method to RowView\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/10c1c984a138a78f29b468ad50bbd365fd247c2d"}]}],"author":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-06-23 12:52:23.000000000","tz":120},"committer":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-06-24 14:09:13.000000000","tz":120},"subject":"Don\u0027t call frozen_row if there are no matching events","message":"Don\u0027t call frozen_row if there are no matching events\n\nThe ovs_idl.RowEventHandler currently calls frozen_row to snapshot rows\nduring event handling even if there is no matching event.\nThis is not a problem in most cases as neutron already wraps this with\nits OvnDbNotifyHandler which similarly does the matching before calling\nfrozen_row. Additionally the hash ring mechanism on the neutron side\nspreads events across multiple workers.\n\nThe neutron maintenance worker does however use the RowEventHandler\ndirectly for its northbound and southbound IDL and also does not\nparticipate in the hash ring. This causes it to process all events and\nclone all rows from update notifications. In larger deployments this\ncauses the maintenance task to spend the majority of its time inside\nfrozen_row. This in turn can cause issues with disconnects due to\ninactivity probe timeouts and transactions timing out on the neutron\nside as the maintenance worker struggles to keep up with event\nprocessing.\n\nAdd the necessary filtering right here on the RowEventHandler instead\nof patching users on the neutron side individually.\n\nChange-Id: Ic8c098d22f3cfc3fadd7f7e1699c1f9a7ea57c9f\nSigned-off-by: Felix Moebius \u003cfelix.moebius@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b51e6cb04122541598af4e2cad766ac05ae32b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b51e6cb04122541598af4e2cad766ac05ae32b46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
