)]}'
{"id":"openstack%2Fpanko~654319","triplet_id":"openstack%2Fpanko~master~Icd2e1a266bb0e26a059bad8aa60fb94dd9f261f7","project":"openstack/panko","branch":"master","topic":"fix_install_command","hashtags":[],"change_id":"Icd2e1a266bb0e26a059bad8aa60fb94dd9f261f7","subject":"Fix install_command in tox.ini","status":"MERGED","created":"2019-04-22 06:28:52.000000000","updated":"2019-04-22 08:33:37.000000000","submitted":"2019-04-22 08:33:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654319-1555922017700-b16e9fe7","meta_rev_id":"01d093ac3efdbad65abe1fffabf4b9403f1e6d5b","_number":654319,"virtual_id_number":654319,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-04-22 08:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-04-22 07:32:14.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-04-22 07:32:14.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 07:32:14.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-04-22 08:33:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4337558f31ed5457b8c00e21721fbd7a9e71c3bd","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-22 06:28:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"893c7f726da0286318b68baf48ccdf662d72cb98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 07:30:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/19/654319/1/check/openstack-tox-pep8/1a5aa1c/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/19/654319/1/check/openstack-tox-py27/5f001ee/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/19/654319/1/check/openstack-tox-py36/1cff1e9/ : SUCCESS in 5m 41s\n- openstack-tox-py37 http://logs.openstack.org/19/654319/1/check/openstack-tox-py37/338f392/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/19/654319/1/check/openstack-tox-docs/2ac9f63/html/ : SUCCESS in 4m 15s\n- build-openstack-releasenotes http://logs.openstack.org/19/654319/1/check/build-openstack-releasenotes/ca86f92/html/ : SUCCESS in 6m 22s\n- telemetry-dsvm-integration http://logs.openstack.org/19/654319/1/check/telemetry-dsvm-integration/98a48f9/ : SUCCESS in 50m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d7871d98c5ba274ddd5998ea307dbd7a44aa8408","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-04-22 07:32:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e1277280b9a82e025aca70545fa99b9991a761a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 07:32:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c59ec023cf457a91b5aa5afabcd554ca5804cb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:33:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eccd0f66269f15cf4889aa9024e800f06188766d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:33:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/19/654319/1/gate/openstack-tox-pep8/a91cc6e/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/19/654319/1/gate/openstack-tox-py27/e637dac/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/19/654319/1/gate/openstack-tox-py36/032ce34/ : SUCCESS in 5m 10s\n- openstack-tox-py37 http://logs.openstack.org/19/654319/1/gate/openstack-tox-py37/bcf4d17/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/19/654319/1/gate/openstack-tox-docs/bd1335d/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes http://logs.openstack.org/19/654319/1/gate/build-openstack-releasenotes/b6a2357/html/ : SUCCESS in 5m 57s\n- telemetry-dsvm-integration http://logs.openstack.org/19/654319/1/gate/telemetry-dsvm-integration/21e24fa/ : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"852d853e21c223ed87f47622f9aa799e5333acf4","revisions":{"852d853e21c223ed87f47622f9aa799e5333acf4":{"kind":"REWORK","_number":1,"created":"2019-04-22 06:28:52.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/19/654319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/19/654319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/19/654319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/19/654319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/19/654319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/19/654319/1"}}},"commit":{"parents":[{"commit":"43f03c9f427c4e06070a0b922e11d7d18a932671","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/43f03c9f427c4e06070a0b922e11d7d18a932671"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 06:28:36.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 06:28:36.000000000","tz":480},"subject":"Fix install_command in tox.ini","message":"Fix install_command in tox.ini\n\nChange-Id: Icd2e1a266bb0e26a059bad8aa60fb94dd9f261f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/852d853e21c223ed87f47622f9aa799e5333acf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/852d853e21c223ed87f47622f9aa799e5333acf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
