)]}'
{"id":"openstack%2Fpanko~680371","triplet_id":"openstack%2Fpanko~master~I94bc8529437ab04cf5372059ef6509b36a7b5e49","project":"openstack/panko","branch":"master","topic":"ipv6-only-deployment-and-testing","hashtags":[],"change_id":"I94bc8529437ab04cf5372059ef6509b36a7b5e49","subject":"Run \u0027telemetry-dsvm-integration-ipv6-only\u0027 job in gate","status":"MERGED","created":"2019-09-05 11:10:00.000000000","updated":"2019-09-06 01:44:00.000000000","submitted":"2019-09-06 01:40:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680371-1567734025721-72f0986f","meta_rev_id":"924723e816fe4372ceb2d91a24ad46055ae80bdf","_number":680371,"virtual_id_number":680371,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-09-06 01:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-09-05 11:55:49.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-09-06 00:37:37.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 00:37:37.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-09-06 01:40:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a1a38d9ccf1a7f83937770a09a1bba436bc7240","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-05 11:10:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac69fc8d1ada9db4ffe4c6174640f95cf6409426","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-05 11:10:48.000000000","message":"Topic set to ipv6-only-deployment-and-testing","accounts_in_message":[],"_revision_number":1},{"id":"9a06c6d70861bdb285bcbb95d3b006c2371c7739","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-09-05 11:55:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04f3908d29d079aa6381ca40108ae279e5db1681","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 13:24:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d587214c9a34136af5d7e7a1160df9b : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4828a66a2b8e4933b1ecff8a7d3f7f5a : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4846b48692024d9cb57a90bfb3e2f2f9 : SUCCESS in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bba9345151e4a7aa5ef532da3b4d2ca : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4480e3d442f4ce3bcb29493341af171 : SUCCESS in 6m 05s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8629840d1bf4493f8c97cb5f6bc98805 : SUCCESS in 53m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/447ff38e0ff642cea2502b33d3320318 : SUCCESS in 49m 22s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/33367250618d48b0bbf72179c354225e : SUCCESS in 2h 06m 41s (non-voting)\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/9ec905eba297405180890ceb9de39508 : SUCCESS in 1h 45m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b50aad3451f6cc64ef8c5ee8ecc9e10fee583da4","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-09-06 00:37:37.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"440a530056ef610ca644f2d195b6121a316de1f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 00:37:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01312a69fa6b0c131b1dab4bd477239a76200f90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 01:40:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba28cb5b5214b0aa5f23e19c4f95cb8 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e1565ee4872a4db8bc0b1ee1304a5cb5 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2edff7e032644e298fab104ede7d6fb5 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/daca52b73d6a42e7acdea2ab8d86bb3b : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8724cc418048463b859fdb37cc1aeb54 : SUCCESS in 4m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f0ae047bc9534664892eab3e9d825012 : SUCCESS in 57m 12s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3a0d889bd9be45fc9d618a4b95d73165 : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f06bba36d8847b013d8861b24dd920c89c55533b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 01:40:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"924723e816fe4372ceb2d91a24ad46055ae80bdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 01:44:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ff0702929f48139e674bd79bc21990 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12377d0e9bc7a9afdafe9ebcc0651b3e9b6ce96e","revisions":{"12377d0e9bc7a9afdafe9ebcc0651b3e9b6ce96e":{"kind":"REWORK","_number":1,"created":"2019-09-05 11:10:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/680371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/71/680371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/71/680371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/71/680371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/71/680371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/71/680371/1"}}},"commit":{"parents":[{"commit":"07f2d20b165fa3fd52236e69f1af6bd724b010ae","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/07f2d20b165fa3fd52236e69f1af6bd724b010ae"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-05 11:09:14.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-05 11:09:58.000000000","tz":0},"subject":"Run \u0027telemetry-dsvm-integration-ipv6-only\u0027 job in gate","message":"Run \u0027telemetry-dsvm-integration-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit Run \u0027telemetry-dsvm-integration-ipv6-only\u0027 job in Panko gate.\n\nStory: #2005477\nTask: #35939\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I94bc8529437ab04cf5372059ef6509b36a7b5e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/12377d0e9bc7a9afdafe9ebcc0651b3e9b6ce96e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/12377d0e9bc7a9afdafe9ebcc0651b3e9b6ce96e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
