)]}'
{"id":"openstack%2Fpanko~696477","triplet_id":"openstack%2Fpanko~master~I6b1364954c231a477d493d4a518c080f87058123","project":"openstack/panko","branch":"master","hashtags":[],"change_id":"I6b1364954c231a477d493d4a518c080f87058123","subject":"Fix error config registration of elasticsearch","status":"MERGED","created":"2019-11-28 08:43:52.000000000","updated":"2019-11-29 06:01:39.000000000","submitted":"2019-11-29 06:00:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696477-1575007203502-7789a045","meta_rev_id":"cde47555ba7e80c87a6d732c7ae23c43136a0d34","_number":696477,"virtual_id_number":696477,"owner":{"_account_id":30679,"name":"yuanrunsen","email":"yuanrunsen@unitedstack.com","username":"yuanrunsen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-29 06:00:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":2,"date":"2019-11-29 05:16:03.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-11-28 20:14:20.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"date":"2019-11-28 10:00:57.000000000","_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":1,"date":"2019-11-29 05:16:03.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"date":"2019-11-28 08:48:55.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-28 08:48:55.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-11-28 10:00:57.000000000","updated_by":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"reviewer":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"state":"REVIEWER"},{"updated":"2019-11-28 20:14:20.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-11-29 05:16:03.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-11-29 06:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1529fe63c4792094b3eb5a0fec169551245e3593","author":{"_account_id":30679,"name":"yuanrunsen","email":"yuanrunsen@unitedstack.com","username":"yuanrunsen"},"date":"2019-11-28 08:43:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66476ba3851b86606550cfc41aec3c592e8137f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 10:51:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f1c8ca866d4e82858b2043ce3d77dc : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27071040be704495809ad05ec4dce5b0 : SUCCESS in 10m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bca8251f19ca4dab9c4e3140088add20 : SUCCESS in 10m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fab41000cf774ae98be14d3f8acf26d0 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20b369d4b9c4619a658a703970a1d5c : SUCCESS in 11m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/646fbda007ca48139daf8c1aab98c8dd : SUCCESS in 1h 06m 39s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b79f94f9cee7442da7b9fb4d5f1bac54 : SUCCESS in 46m 17s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/1fd672313bf545aba359366be5bae984 : SUCCESS in 2h 00m 32s (non-voting)\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4e393e60f4d74f56993af1e9958914d6 : SUCCESS in 2h 01m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a62e16ae430e3a795b5721de88ae696439f5aed","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-11-28 20:14:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49a8f6cd41b3deb7415d3ab45812693985e4c64e","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-11-29 05:16:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"44e4e005e8bc5783afa456f2eb4fd71de1cb3d81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 05:16:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f43140d590acf5f668bab0ea6a415b2f897577d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 06:00:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d2639ddd7459d6bd74a9b68330f43ae780cdeb20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 06:00:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86fcb82fc95840229077a8ba468a982a : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e26abca6033488dacff07c2a7ad3229 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfe0f6b8006945f0933684f1caac071d : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb039f4c631147dcb40b6c334f8bbca7 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7dfbb4bdeb846adb3ff5815023032e5 : SUCCESS in 5m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a4aca883b2d047b0ba0ac355726194d9 : SUCCESS in 36m 42s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2c3056fa9b2f46d28c440956dd53ffb9 : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":1},{"id":"cde47555ba7e80c87a6d732c7ae23c43136a0d34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 06:01:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f8dc1eff144e66adb2a923ac0b870b : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9e8be6cc0a7bdc06beaba4f7c3d168fee16a634","revisions":{"e9e8be6cc0a7bdc06beaba4f7c3d168fee16a634":{"kind":"REWORK","_number":1,"created":"2019-11-28 08:43:52.000000000","uploader":{"_account_id":30679,"name":"yuanrunsen","email":"yuanrunsen@unitedstack.com","username":"yuanrunsen"},"ref":"refs/changes/77/696477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/77/696477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/77/696477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/77/696477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/77/696477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/77/696477/1"}}},"commit":{"parents":[{"commit":"59245276791dcc275aec10aa002ccd645d1afd69","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/59245276791dcc275aec10aa002ccd645d1afd69"}]}],"author":{"name":"yuanrunsen","email":"yuanrunsen@unitedstack.com","date":"2019-11-28 07:15:41.000000000","tz":0},"committer":{"name":"yuanrunsen","email":"yuanrunsen@unitedstack.com","date":"2019-11-28 08:43:26.000000000","tz":0},"subject":"Fix error config registration of elasticsearch","message":"Fix error config registration of elasticsearch\n\nConfiguration options must be registered before using them.\nIn impl_elasticsearch.py, using conf.database.es_index_name\nor using conf.database.es_index_name are all wrong that will\nlead the api server to be invalid.\n\nChange-Id: I6b1364954c231a477d493d4a518c080f87058123\nSigned-off-by: yuanrunsen \u003cyuanrunsen@unitedstack.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/e9e8be6cc0a7bdc06beaba4f7c3d168fee16a634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/e9e8be6cc0a7bdc06beaba4f7c3d168fee16a634"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
