)]}'
{"id":"openstack%2Fpanko~699330","triplet_id":"openstack%2Fpanko~master~I028e426db2a98b2377b58b0a184868307d036438","project":"openstack/panko","branch":"master","hashtags":[],"change_id":"I028e426db2a98b2377b58b0a184868307d036438","subject":"add brief introduction of this project","status":"ABANDONED","created":"2019-12-17 03:15:51.000000000","updated":"2019-12-17 06:26:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"23ae4fa41fd50df88c52329082a9728150091e95","_number":699330,"virtual_id_number":699330,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-17 03:26:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":-2,"date":"2019-12-17 05:56:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 03:26:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-17 05:56:33.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"394b2fd9309c9ae0c3f5166881dd8721a3274df8","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2019-12-17 03:15:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d83de20106b6aae4cb2391301074b45740cc221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 03:26:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9e453f90439a47898bd337ddf6392410 : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e17a95c4204e4a2bb7b0eb0c22d349f6 : FAILURE in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae0c2f1c58e47fabdbce508266208d1 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"90e254c20435553fa841aea47e1aaae99dd88487","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-17 05:56:33.000000000","message":"Patch Set 1: Code-Review-2\n\nProject description can be found here https://docs.openstack.org/panko/latest/","accounts_in_message":[],"_revision_number":1},{"id":"23ae4fa41fd50df88c52329082a9728150091e95","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2019-12-17 06:26:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8a226a337113ff9730555d6cace0cc7a212a96a","revisions":{"b8a226a337113ff9730555d6cace0cc7a212a96a":{"kind":"REWORK","_number":1,"created":"2019-12-17 03:15:51.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/30/699330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/30/699330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/30/699330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/30/699330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/30/699330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/30/699330/1"}}},"commit":{"parents":[{"commit":"b787e1dc4da06dd0d05ef57e7799cde9741307c1","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/b787e1dc4da06dd0d05ef57e7799cde9741307c1"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2019-12-17 03:13:29.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2019-12-17 03:13:29.000000000","tz":480},"subject":"add brief introduction of this project","message":"add brief introduction of this project\n\nChange-Id: I028e426db2a98b2377b58b0a184868307d036438\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/b8a226a337113ff9730555d6cace0cc7a212a96a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/b8a226a337113ff9730555d6cace0cc7a212a96a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
