)]}'
{"id":"openstack%2Fpanko~721172","triplet_id":"openstack%2Fpanko~master~I470b83580d3f8040baf0cf1d4cf964713f151bdb","project":"openstack/panko","branch":"master","hashtags":[],"change_id":"I470b83580d3f8040baf0cf1d4cf964713f151bdb","subject":"Add irrelevant-files for tempest jobs","status":"MERGED","created":"2020-04-20 06:35:34.000000000","updated":"2020-04-20 11:51:28.000000000","submitted":"2020-04-20 11:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721172-1587383129623-a68383b8","meta_rev_id":"d2fb8c6cc60ac4e300226dbfe5ec2471a6eb0406","_number":721172,"virtual_id_number":721172,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 11:45:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-04-20 10:55:46.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-04-20 10:55:46.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 11:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6ffcce1e412b59c6093ee1146e959917e231069","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-20 06:35:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8ca7a8fe4b51801e3e1dcbf2f7fcf48724d19f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 07:26:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946d805ec250484e91d4ae561dbe49ad : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8d9ac248b840b6b106227a01876bb0 : SUCCESS in 10m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66906cdd16b84c3ba74f2b419b1b0893 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0a61782e3204f1582e29d5e002ebeff : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d19959c734448bbe4cd46ad74b9f4a : SUCCESS in 11m 40s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2701c1b120e6418fa455b3fa8d1aae1e : SUCCESS in 47m 47s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2b7669435db64bc4b3dc9c2c7411274a : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":1},{"id":"cdc44b64fc20835d5e23149e7a581d33e4ef41d3","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-20 10:55:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cc13ececad0b18b47baf5bff918de9693535e9a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 10:56:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff110fc7eeede349b755b7119b625106c9a35094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 11:45:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91cd852942984530aa0160b7584fc13a : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/832f11825f9947cdabf36b5695560f81 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65db1465393e44f58e36ac6a3aa4dab6 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22931591649462fb3e0614b4fd51711 : SUCCESS in 7m 41s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7261772090f449529df69fdf2c8b07df : SUCCESS in 44m 07s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1f01239a3d9841eea866e270af20b63b : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9809e7908e0aaa2adbf61f2ae0531ae60461b38f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 11:45:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d2fb8c6cc60ac4e300226dbfe5ec2471a6eb0406","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 11:51:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4593583724574edc90c74e9b1af1bddb : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ee7cbcd25e6d9214ed7043594f857693a484ec2","revisions":{"5ee7cbcd25e6d9214ed7043594f857693a484ec2":{"kind":"REWORK","_number":1,"created":"2020-04-20 06:35:34.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/72/721172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/72/721172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/72/721172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/72/721172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/72/721172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/72/721172/1"}}},"commit":{"parents":[{"commit":"679e39587d3895d951954b084dd69816f79ef3e9","subject":"Cleanup doc support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/679e39587d3895d951954b084dd69816f79ef3e9"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-20 06:35:13.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-20 06:35:13.000000000","tz":-420},"subject":"Add irrelevant-files for tempest jobs","message":"Add irrelevant-files for tempest jobs\n\nChange-Id: I470b83580d3f8040baf0cf1d4cf964713f151bdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/5ee7cbcd25e6d9214ed7043594f857693a484ec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/5ee7cbcd25e6d9214ed7043594f857693a484ec2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
