)]}'
{"id":"openstack%2Fpanko~722738","triplet_id":"openstack%2Fpanko~master~Ied79c449a8bd0c53a7392afb1dee18b5e7e5d619","project":"openstack/panko","branch":"master","topic":"add-py38-meta","hashtags":[],"change_id":"Ied79c449a8bd0c53a7392afb1dee18b5e7e5d619","subject":"Add py38 package metadata","status":"MERGED","created":"2020-04-24 13:26:08.000000000","updated":"2020-04-26 01:04:15.000000000","submitted":"2020-04-26 01:02:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722738-1587862974778-81340386","meta_rev_id":"0ce935da92694f14dde547bdcbed0470e338fade","_number":722738,"virtual_id_number":722738,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-26 01:02:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-04-26 00:51:57.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-04-26 00:51:57.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 00:51:57.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-04-26 01:02:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c30d4426b9b49759bc52d7737a5e114d0e4a40fb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 13:26:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c37c1376824f9801d3174f7d6fb75fae6f1fb4ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:47:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0aeb92bf97d4490e917633d6a66abe4a : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ef954c009f49e4975362f4882a486c : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5d93f8c473e46999028d0466915be8e : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74e985bf615941059ba6d93246af9856 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2254a411ca9409fa265352672529897 : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3cb91aa08b7767d4cbc576a3dfff446180637bf9","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-26 00:51:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9471ea0d003a1eb19400b41e348a9c62d302a47a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 00:52:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39aa222fefced47e91002f053d0879a732fcb651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 01:02:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/804ed2d1e1124be8b7d58e00aa2d59ca : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d10fbfdf30b45bbbe84def19f02ee8a : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd68ef6baa6f468fbde0906d2bb26492 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42f97c4c24784af59a8b15f9857c5fc5 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b428f53d086941f6ba89f0cce36cf7c9 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f6d54971aa8c134a96622312c37ecfaa7231d853","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 01:02:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"230179796af1f1dc4bc4959f76d527963f139a1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 01:04:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f793978b844f9b87f8b8e4c7d66c36 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8583d1a6989203ffce02568b9d7c0e7a5c9a8b6c","revisions":{"8583d1a6989203ffce02568b9d7c0e7a5c9a8b6c":{"kind":"REWORK","_number":1,"created":"2020-04-24 13:26:08.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/722738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/38/722738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/38/722738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/38/722738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/38/722738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/38/722738/1"}}},"commit":{"parents":[{"commit":"ba63796b204d60b7fc4d1199f9d5c23b22326e4d","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ba63796b204d60b7fc4d1199f9d5c23b22326e4d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:23:15.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:23:15.000000000","tz":-300},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nNow that we are running the Victoria tests that include a\nvoting py38, we can now add the Python 3.8 metadata to the\npackage information to reflect that support.\n\nChange-Id: Ied79c449a8bd0c53a7392afb1dee18b5e7e5d619\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/8583d1a6989203ffce02568b9d7c0e7a5c9a8b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/8583d1a6989203ffce02568b9d7c0e7a5c9a8b6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
