)]}'
{"id":"openstack%2Fpanko~729470","triplet_id":"openstack%2Fpanko~master~Ic5bd9d01030fd337fc47f94c8abc28793622abc1","project":"openstack/panko","branch":"master","topic":"fuzzy-query","hashtags":[],"change_id":"Ic5bd9d01030fd337fc47f94c8abc28793622abc1","subject":"Make events support fuzzy query based on event type","status":"NEW","created":"2020-05-20 01:49:02.000000000","updated":"2020-06-23 10:47:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a5d189cc59ef30fd1cbce42e9b1f7068e43caad6","_number":729470,"virtual_id_number":729470,"owner":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-23 10:47:43.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-20 02:47:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 02:47:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-23 10:47:43.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"}],"messages":[{"id":"8b759919c7bd228bab01d67340ba2f1305346fed","author":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"date":"2020-05-20 01:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6f39b7900c954f458a8c9dd36464baa1bd7ac67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 02:47:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ebe56f0a28544309c6a7184c82ed538 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b753aa3cdd704271841d2d6dc549579a : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d75192ce29e14186a0176fffd4bd5364 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e75a43d5025749fcb9050388e7d43556 : SUCCESS in 8m 27s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d170677cb7614fccac64600078d3a00b : SUCCESS in 47m 17s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d685da53760143f19b789da05d500195 : SUCCESS in 52m 33s","accounts_in_message":[],"_revision_number":1},{"id":"113e8511f0c8a10e636c0d80c5bcd4ff6205ca56","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-23 10:47:43.000000000","message":"Patch Set 1:\n\nI\u0027m not feeling comfortable to put -1 on this patch because I\u0027m not involved a lot on it, but this change changes API so it\u0027s better to have at least release note for it","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11c7e7ce475d3b0ef0adb0a6af2260e94be33669","revisions":{"11c7e7ce475d3b0ef0adb0a6af2260e94be33669":{"kind":"REWORK","_number":1,"created":"2020-05-20 01:49:02.000000000","uploader":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"ref":"refs/changes/70/729470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/70/729470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/70/729470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/70/729470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/70/729470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/70/729470/1"}}},"commit":{"parents":[{"commit":"8583d1a6989203ffce02568b9d7c0e7a5c9a8b6c","subject":"Add py38 package metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/8583d1a6989203ffce02568b9d7c0e7a5c9a8b6c"}]}],"author":{"name":"guozijian","email":"guozijian@unitedstack.com","date":"2020-05-20 01:48:14.000000000","tz":480},"committer":{"name":"guozijian","email":"guozijian@unitedstack.com","date":"2020-05-20 01:48:55.000000000","tz":480},"subject":"Make events support fuzzy query based on event type","message":"Make events support fuzzy query based on event type\n\nChange-Id: Ic5bd9d01030fd337fc47f94c8abc28793622abc1\nSigned-off-by: guozijian \u003cguozijian@unitedstack.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/11c7e7ce475d3b0ef0adb0a6af2260e94be33669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/11c7e7ce475d3b0ef0adb0a6af2260e94be33669"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
