)]}'
{"id":"openstack%2Fpanko~729978","triplet_id":"openstack%2Fpanko~master~I8d4234212ecceb639e40eae423af72d2d08b010d","project":"openstack/panko","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"I8d4234212ecceb639e40eae423af72d2d08b010d","subject":"Switch to newer openstackdocstheme and reno versions","status":"MERGED","created":"2020-05-21 14:20:18.000000000","updated":"2020-05-26 01:39:20.000000000","submitted":"2020-05-26 01:38:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729978-1590457087418-4b0285ba","meta_rev_id":"29f66a1ddc645dcace2451a3b43c780fbaac0c62","_number":729978,"virtual_id_number":729978,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 01:38:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-05-26 01:26:44.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-05-26 01:26:44.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 01:26:44.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-05-26 01:38:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06b784cfce739174c7da5bdb181aa9674dfd2c9a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-21 14:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e93d1057a3f50447b0b57f67ac3f8ed024f9f299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 14:35:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1684109f0dc24c50b633f9daf78a9460 : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39d7db5c729c4df4a857c50e074b4e92 : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeb11a79f01d4dc893b90f83d6d26f1d : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0842b57be7d54e7bb641ce297e0d1223 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/179fc90d10714388be42a9300a9b0dd8 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8042d35303aad2626269d6d22733a5f0370433d1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-05-26 01:26:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e2076b3f921fb19c62da869c757541865a3f618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 01:26:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c4d71b0f4e2101b8c1ef433f121e7c697f14bd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 01:38:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ba1037370c94720a86322bf0186519e : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7226e2d5472743e58523c88f39775b50 : SUCCESS in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dc7e0109592467d93618ae3696ec60f : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c24b504dafae44aebd49c68ebef4bc05 : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2864686fa1604f588fe70d7ccc0507a7 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"07c94190eea5fc243b04fe960f8658346882437f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 01:38:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e54e10a19c6cd8cb1db79b13251ae85e46186b59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 01:39:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3f41fda49a4f22856fab08232818cd : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d57390f044114979a0043f32fb946914 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","revisions":{"260ba07210c3474afe20b3cf4fe42cfbbe905f9e":{"kind":"REWORK","_number":1,"created":"2020-05-21 14:20:18.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/78/729978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/78/729978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/78/729978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/78/729978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/78/729978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/78/729978/1"}}},"commit":{"parents":[{"commit":"8583d1a6989203ffce02568b9d7c0e7a5c9a8b6c","subject":"Add py38 package metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/8583d1a6989203ffce02568b9d7c0e7a5c9a8b6c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-21 14:18:39.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-21 14:20:05.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nDisable openstackdocs_auto_name to use \u0027project\u0027 variable as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nRemove docs requirements from test-requirement, it\u0027s enough to have them\nin doc/requirements.txt.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I8d4234212ecceb639e40eae423af72d2d08b010d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
