)]}'
{"id":"openstack%2Fpanko~733101","triplet_id":"openstack%2Fpanko~master~Icf83ffe089301b3782273923f18efd4d209131c2","project":"openstack/panko","branch":"master","topic":"expire","hashtags":[],"change_id":"Icf83ffe089301b3782273923f18efd4d209131c2","subject":"Add max_count argument for clear_expired_data call","status":"MERGED","created":"2020-06-03 11:11:52.000000000","updated":"2020-09-15 13:31:30.000000000","submitted":"2020-07-02 12:49:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733101-1593694195038-ea166436","meta_rev_id":"5ee30ded4c78867e8c6b134fe7d0bd196d942c81","_number":733101,"virtual_id_number":733101,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 12:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-02 11:50:23.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-02 11:50:23.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 10:16:41.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2020-06-23 15:16:09.000000000","updated_by":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"reviewer":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"state":"REVIEWER"},{"updated":"2020-06-26 07:21:03.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-06-29 08:44:47.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-07-02 11:50:23.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-07-02 12:49:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecf56125773496ee897a3c160ed4623443525c48","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-03 11:11:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6761d0cfded8eba81e03bd9b391bb3cd90e05f12","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-03 11:13:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10960050c5dc5dbf20b27173a98c331f352090fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:14:27.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e5c2ede783407190a34de8a3795ee0 : FAILURE in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1f8eb3e7e964d70b335d0498b629036 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9631b8b87f648379ec426fe427c718b : SUCCESS in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b16cf1bac9343c5b445e51d45f5d81d : SUCCESS in 13m 17s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9645bb0948c8493e97928f876b2d1a79 : SUCCESS in 46m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c2371dfdcadf4dd3a1b7f2071ea450f3 : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e274af8f71f0e7743df1887b9c87508cb46de2de","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-04 10:28:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9544e5c5b1f40cd8a0ca24265f2b94ceebe6ac9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:27:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb2b72ced7c5409abe33da9a35807279 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2330e9fee03644dbad036d1c07bc518d : FAILURE in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fe5190958384c0a8243c6c35937f287 : FAILURE in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b64f427b9d4683bbe5507cabfdf60c : SUCCESS in 14m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a5956d7b0de44efb7a94c485b79374c : SUCCESS in 8m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e2cbaec8ba684edaa568e37f28f191f6 : SUCCESS in 45m 33s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/24d0ad8e34054d44b937db59029203f8 : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":3},{"id":"19fe8a9d050f12ab653cc6760bdb36f265a1ac07","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-05 13:32:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e57fba4887832b3585f6ae89ce350c8b68a2804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 14:30:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83de67ff7c9a4805a17505e65dcc27e2 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdf2da102f03409396f568192bfb1bff : FAILURE in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/178e1e301d8c4030956bacac323d66f4 : FAILURE in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f996eb7bf454cf4a5e2002fa921125d : SUCCESS in 7m 15s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3a02c8677c304f7cb00b70ccc33c5dfe : SUCCESS in 54m 25s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2dd7cf7120024f46ae6aae4e46becaed : FAILURE in 55m 46s","accounts_in_message":[],"_revision_number":4},{"id":"ae6c9a77372ff6afc1e7508b9a227a68916933e1","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-06 08:31:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"33a7b2abf56e83a59091aab972f73b49568e3784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 09:19:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d4a91e84b2481b880a28bee5b59c26 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37037ebaf18e44009803b907446ba739 : SUCCESS in 10m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01daa5e21d934869bfac8e895fffd1b4 : SUCCESS in 9m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30abeb5eea9048e997fa5c68b80cf83a : SUCCESS in 9m 17s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/39b0c4339c7449aa82195e07e9d554a8 : SUCCESS in 46m 41s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/69c19fee14734e3db143d484493c2841 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":5},{"id":"ed1b8ee221cb2cefd1b2513ede869f9265d6d1e4","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-06-13 01:33:15.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks, Ivan, some minor comments","accounts_in_message":[],"_revision_number":5},{"id":"cef17096002964e1ec08796de20640f10c69d47c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-15 09:30:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"469795c3c0681bbe78a6d527a9ec94beadc895fc","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-15 09:30:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6a715c7abb2b0895d0f4cb82e2ec15f757ad8ece","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-06-15 09:43:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b657363a1a407e817e21f6846fe2ba21c1422945","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-15 09:55:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"af216f8a2377ac3b9465e52a385c29b854ae59c4","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-06-15 10:43:51.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nIt seems with this patch you are breaking default behaviour:\n- before all events with TTL expired were deleted\n- now only 100 such events are deleted by default, and operator has to repeat this command again and again (with no actual feedback on when it is actually ready).\n\nI\u0027d suggest to rename the option to hint on \u0027batch size\u0027 and just always delete the events in batches, with configurable batch size. no need for special meaning for \u003c\u003d0. This way only performance/stability characteristics of this operation will change, not the flow of its usage by cloud operators","accounts_in_message":[],"_revision_number":7},{"id":"70d173ede44072c3ac2bdf7465109ba819f63f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 10:57:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e14811b8f794c94a77df0199c49f37b : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e78d92899f5f47cca6d01a6e604e80b3 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d62f577107844752823ce3ba84cfcc75 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1ab9d88148427a9de4743adb251d09 : SUCCESS in 7m 33s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6495cf0b65ef4672a825dae6ecdd3d10 : SUCCESS in 51m 59s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/72bac94ef5ee411fa040c0609287344b : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":7},{"id":"703e974184881a8aa946a9e54a3b2995201dce04","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-15 14:16:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2e4110b1595a0165dfd9404c7e58a5acda718646","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-15 14:17:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f4ef88a08307e5621f63482309a4915774375546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 15:21:57.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65d5fe97b17141e8ac63687227d98e00 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18bf88593d5c4030b6d60333d35856f4 : FAILURE in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6cc0ef6c64c4feaa3ad23880ac79a4a : FAILURE in 12m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812db507eb2242fa9b4471b3065caa97 : SUCCESS in 10m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1a74226a0c984c70981f5363f8d2b181 : SUCCESS in 54m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/db40bcd232c444428161940ea570d224 : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":8},{"id":"6461e6ed4dfaa66a6a9526fc59cc4db59a970dee","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-15 15:57:25.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"b67baac6d3ce054e394806e888a410096f9aa101","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-15 16:34:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b0360003b396d8b633efb424627102c255db7af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 17:46:43.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/895e14a7e85e455693fa86c5ce256fa3 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e623d62e11b54b6c80ea605dd9d58d12 : SUCCESS in 11m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c2964aef0ba41fdac65064e956e00a4 : SUCCESS in 15m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a3f18a3e9934f92948c94e3ec51baa5 : SUCCESS in 17m 37s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e6f52835515c4e59a6d14360d483bf2a : FAILURE in 1h 03m 03s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/57a8bb0749bd49a5a5939eabfcd7ebc2 : FAILURE in 1h 04m 20s","accounts_in_message":[],"_revision_number":10},{"id":"e31a835501203944468b904bd321663317df9029","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-15 21:55:25.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"fbcf9764f10cccec8e08ad6a3340ad62f8196e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 23:00:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0cedcefc91b473d9dc14e651ffad5a1 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cecf0211f5e48e385a518c61e79c034 : SUCCESS in 9m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/358f6cb63b96465ba714d06ba68c2c5f : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/786e0db0854347ca82931774cd073984 : SUCCESS in 12m 48s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ad64fa30ab7a4b878616937d48fe78dd : SUCCESS in 51m 58s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4910dc6dfa5945e3b70d6f39cea08862 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":10},{"id":"75d7d884a849a029f95adb5480da534397ded791","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-06-16 10:16:41.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bad1c1389b40ce07abd6795c25ff476a210a9a6e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-16 10:42:20.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3983b3a1fedadf738781f706f6a80661d58c1fb1","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-06-16 11:28:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"393c9e6cf901803909d3ef219e13f3ee88eae242","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-17 21:25:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8232c81f31e05c1b4162287e0a4ba863f3cd8d58","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2020-06-23 15:16:09.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7167885cd21a05def81cb22085720bdec03be485","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-24 09:56:42.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"9d20bfd0be9cbc227074faa7968c065326888842","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-25 08:18:22.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9bb5588d5f6c0561154a5b0b3adff87a69e240dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 09:24:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02e23bfee10c4afbba3fb98bdb66bcbb : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1646c8050b164a04b58d573868986769 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49cb84cfad9c486b9749bab49f210e9d : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/599f34a946cb4c6f8ff3d944142a1150 : SUCCESS in 9m 54s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3b9fdc4bf5314426bf76b969ddc5d416 : SUCCESS in 1h 01m 12s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3748ac7f85e44689b211353f72d14cf2 : SUCCESS in 53m 55s","accounts_in_message":[],"_revision_number":11},{"id":"d89a0fcb99243e519f72c5197a6171578f639be4","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-06-26 07:21:03.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"d0858e40d333a15c7b10cabb4dea93388add6dfe","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-06-29 08:44:47.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"de42ac347a6fd22e52e53774a41360fd53dca866","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-29 12:45:59.000000000","message":"Patch Set 11: Code-Review+1\n\nlgtm\n\nAs this is an operation-wise change, could you please also add a release note? For the ongoing release note file name in Panko, I\u0027d suggest we use \u0027\u003crelease\u003e-\u003cdescription\u003e.yaml\u0027, e.g. victoria-support-batch-delete-events.yaml","accounts_in_message":[],"_revision_number":11},{"id":"f8c0c7b9723bbeb24c6c0a2d82535fddeb4ca093","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-01 13:09:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9007aa7b8154b8e8d559371ca626c9fb28774838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 14:05:53.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/696e484ec1e14e3386af0579c93f2aa3 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f77a931dc854dc2b1388cff01445145 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e5fe9d4e0b94085aeffe429b2796105 : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f895b3f9fd14fc6bd14f1c2af985c11 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a0a16ed1bf7419cae1d8fa48258337d : SUCCESS in 5m 34s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cbdb376e70ab44a8afb3cf51ab4cab3b : SUCCESS in 48m 04s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/cdae95a5bf6f4bc89e7d03f24c22eff6 : SUCCESS in 46m 02s","accounts_in_message":[],"_revision_number":12},{"id":"ab270cc46dc6c9886230ba4bc8eb0adb9bb26aff","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-01 21:14:55.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"4757d7646207c55a0688aee461c45836069cf121","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-01 21:14:59.000000000","message":"Patch Set 12: Code-Review-1","accounts_in_message":[],"_revision_number":12},{"id":"27750707119a2c0af7031306b0cec921eaf37f92","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-02 07:31:01.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"3eb83551a7256488bf434f4712d29edff7ac4a95","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-02 07:31:17.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"831a4e2072eb366885979d0315a3823ca516b567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:30:05.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ad91b87b3c41e69b02d93fa926b630 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eeac6d9014d4b089562f83286f7a919 : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ac3243089084f9d9008a3e41c064966 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f1dfcd2e924529b9a77c3327a48a4c : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1fd8f7eeba540059903acfde242c868 : SUCCESS in 7m 34s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e5829296860f4a74a266eae5ac04255c : SUCCESS in 49m 13s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/69b28ef3e4104ed0be7a4a9041ef5930 : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":13},{"id":"a459e23731ad806aad8509911bef836318ecbcc9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-02 11:50:23.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\nFast merge this patch given there was already a +2 from zhurong.","accounts_in_message":[],"_revision_number":13},{"id":"a20eb5f2b138fc145b93529bdd8d627fe8d8ae92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 11:50:36.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"21dc36d37d125075a93fddf5f18cff3cf0776ad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:49:54.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb7cceff08d34d3f913a74992ae5d0a3 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06c26340416143dcac445a409420a0a2 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa7b23509ed14b7ea4ebbb3bc24c2849 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f369497942664e9e9e91c8cc1e1ce529 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fafb22430484b53897963f5601b5897 : SUCCESS in 5m 34s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/fdf9ad6dd58748408ac123f1db3e69a0 : SUCCESS in 47m 12s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/946b1bda5b3248d186e76b6c0ebef1e7 : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":13},{"id":"c5aa1f08ff1c7391df4ae76d689b338b19903aab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:49:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"903492ec99360f97f939a34d0f32017179682ce4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:52:14.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0607cc0faa11472a8c12e2a48e75c129 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/899c8c66244442f7a8770557db63fda0 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":13},{"id":"ca026e17ff48b1eef5b3f48a2d14594b6d0a7646","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-15 13:31:30.000000000","message":"Patch Set 13: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 0a8d5b4fc0169108468404e88d12ee6fefed4e63","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"ec84ed8aa7c75d83b50ddac3edb4025462a61084","revisions":{"87acabc0e25096724f10671330da8242d019482d":{"kind":"REWORK","_number":1,"created":"2020-06-03 11:11:52.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/1"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:48.000000000","tz":180},"subject":"WIP. Add batch_size argument for clear_expired_data call","message":"WIP. Add batch_size argument for clear_expired_data call\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/87acabc0e25096724f10671330da8242d019482d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/87acabc0e25096724f10671330da8242d019482d"}]},"branch":"refs/heads/master"},"95857b65fa747554d4db79c1eea7e66602bc05e2":{"kind":"REWORK","_number":2,"created":"2020-06-03 11:13:12.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/2"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:13:00.000000000","tz":180},"subject":"WIP. Add batch_size argument for clear_expired_data call","message":"WIP. Add batch_size argument for clear_expired_data call\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/95857b65fa747554d4db79c1eea7e66602bc05e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/95857b65fa747554d4db79c1eea7e66602bc05e2"}]},"branch":"refs/heads/master"},"e2156dfd074606f2c8cfda9b077583e4f51a1c2c":{"kind":"REWORK","_number":3,"created":"2020-06-04 10:28:01.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/3"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-04 10:27:48.000000000","tz":180},"subject":"WIP. Add batch_size argument for clear_expired_data call","message":"WIP. Add batch_size argument for clear_expired_data call\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/e2156dfd074606f2c8cfda9b077583e4f51a1c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/e2156dfd074606f2c8cfda9b077583e4f51a1c2c"}]},"branch":"refs/heads/master"},"d0cb50fbf2892818a4dac362d602f0c9c6b5dbc8":{"kind":"REWORK","_number":4,"created":"2020-06-05 13:32:03.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/4"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-05 13:32:01.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_to_delete` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/d0cb50fbf2892818a4dac362d602f0c9c6b5dbc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/d0cb50fbf2892818a4dac362d602f0c9c6b5dbc8"}]},"branch":"refs/heads/master"},"05a76a14ea93cc35ed2b1357070fb70620ebbcfe":{"kind":"REWORK","_number":5,"created":"2020-06-06 08:31:11.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/5"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-06 08:31:01.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_to_delete` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/05a76a14ea93cc35ed2b1357070fb70620ebbcfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/05a76a14ea93cc35ed2b1357070fb70620ebbcfe"}]},"branch":"refs/heads/master"},"066c915831b46e03c5de37347ae20e3a1603f8d8":{"kind":"REWORK","_number":6,"created":"2020-06-15 09:30:47.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/6"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-15 09:30:35.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_to_delete` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/066c915831b46e03c5de37347ae20e3a1603f8d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/066c915831b46e03c5de37347ae20e3a1603f8d8"}]},"branch":"refs/heads/master"},"3442be328874cdc7c425822e3d52bb45186cb3a9":{"kind":"REWORK","_number":7,"created":"2020-06-15 09:55:19.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/7"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-15 09:55:08.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_to_delete` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/3442be328874cdc7c425822e3d52bb45186cb3a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/3442be328874cdc7c425822e3d52bb45186cb3a9"}]},"branch":"refs/heads/master"},"ea65f7c843717518cb2f90d4647b78aeda041aa8":{"kind":"REWORK","_number":8,"created":"2020-06-15 14:17:13.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/8"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-15 14:17:02.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_to_delete` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ea65f7c843717518cb2f90d4647b78aeda041aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ea65f7c843717518cb2f90d4647b78aeda041aa8"}]},"branch":"refs/heads/master"},"cca830fcb264e6929bc94be83bf5871a99bdce76":{"kind":"NO_CHANGE","_number":9,"created":"2020-06-15 15:57:25.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/9"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-15 15:57:15.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_to_delete` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/cca830fcb264e6929bc94be83bf5871a99bdce76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/cca830fcb264e6929bc94be83bf5871a99bdce76"}]},"branch":"refs/heads/master"},"0098f1823df1dd1cdcee0cfa1eaed60c31c9b8f7":{"kind":"REWORK","_number":10,"created":"2020-06-15 16:34:31.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/10"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-15 16:34:21.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_to_delete` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/0098f1823df1dd1cdcee0cfa1eaed60c31c9b8f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/0098f1823df1dd1cdcee0cfa1eaed60c31c9b8f7"}]},"branch":"refs/heads/master"},"a66feca95fae2d5ec1bf276951c8fe6f7debb9bd":{"kind":"REWORK","_number":11,"created":"2020-06-25 08:18:22.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/11"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-25 08:18:10.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_to_delete` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/a66feca95fae2d5ec1bf276951c8fe6f7debb9bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/a66feca95fae2d5ec1bf276951c8fe6f7debb9bd"}]},"branch":"refs/heads/master"},"ef0d61c6a839a69be3c0d4f802749d535195dd86":{"kind":"REWORK","_number":12,"created":"2020-07-01 13:09:36.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/12"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-07-01 13:09:10.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_delete_batch_size` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ef0d61c6a839a69be3c0d4f802749d535195dd86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ef0d61c6a839a69be3c0d4f802749d535195dd86"}]},"branch":"refs/heads/master"},"ec84ed8aa7c75d83b50ddac3edb4025462a61084":{"kind":"REWORK","_number":13,"created":"2020-07-02 07:31:17.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/01/733101/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/01/733101/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/01/733101/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/01/733101/13"}}},"commit":{"parents":[{"commit":"260ba07210c3474afe20b3cf4fe42cfbbe905f9e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/260ba07210c3474afe20b3cf4fe42cfbbe905f9e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-07-02 07:31:06.000000000","tz":180},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_delete_batch_size` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ec84ed8aa7c75d83b50ddac3edb4025462a61084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ec84ed8aa7c75d83b50ddac3edb4025462a61084"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
