)]}'
{"id":"openstack%2Fpanko~735867","triplet_id":"openstack%2Fpanko~master~Idf9a184f969e5f77de0c2f2a2b99c2931f5647e0","project":"openstack/panko","branch":"master","hashtags":[],"change_id":"Idf9a184f969e5f77de0c2f2a2b99c2931f5647e0","subject":"Add a /healthcheck by default","status":"MERGED","created":"2020-06-16 10:48:28.000000000","updated":"2020-12-01 15:23:43.000000000","submitted":"2020-12-01 15:18:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735867","meta_rev_id":"37ee9fc555b3f5021dc20d7dbe69485c4339b2c3","_number":735867,"virtual_id_number":735867,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 15:18:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-12-01 15:05:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":1,"date":"2020-07-03 04:08:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-12-01 15:05:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 10:45:23.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-07-02 10:27:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-03 04:08:44.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-12-01 15:05:57.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"c5eb9067d8229f7def413d94fc839485a6e6fabf","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-06-16 10:48:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51ffa19ecb53993e38e018976c60d13ce6212bc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:57:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4501734bed4a4391c09fc1f8d68955 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac3098f85f1248c78e84abed8f2dff1c : SUCCESS in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72c9221faa9040838271a7f39deaeb2a : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe39f89f5874187a5d26844bf97e6ea : SUCCESS in 11m 41s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/63c3f56ece3c437f99312409f84965a0 : SUCCESS in 52m 02s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f2625213545c4c05a77c077a5968605e : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":1},{"id":"46416240adb983cd3270a99fe8c00f6e094bef8d","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-23 10:45:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8db472422fe219509a932c63ca6757000a64c0c","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-07-02 09:27:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62a000cfc6fa4535409bc31d31fc33669346b743","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-07-02 09:27:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"66aa0883991cc49e2cbec54d12c59447c9de52eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 10:27:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c392e12069646159c48a63e1f01b78d : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f4fafd5d51742e0b2c8e9eae16b94ec : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/440fdd82cf9a4be1bb8c13a5383b9bec : SUCCESS in 21m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40f33f0921442c4864fadb93a3cdfd1 : SUCCESS in 9m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/01b519494e5c4d4fb46e3758aefb4499 : SUCCESS in 53m 05s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/5d6f54a9b5704c37acbcddef784731fb : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":2},{"id":"47d7809a262b7ee56b43c1a80101a64632c6f912","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-03 04:08:44.000000000","message":"Patch Set 2: Code-Review+1\n\nrelease note please","accounts_in_message":[],"_revision_number":2},{"id":"0f4ee060f2f7f10a51d4200dcc8b5381d3984cdd","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-01 15:05:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThank you, this looks good to me.","accounts_in_message":[],"_revision_number":2},{"id":"694845fd87366565d98d4f71d519c8d511b1f217","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:06:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3c7c112f55a72fb77bb5e38d677b6309fb6ea812","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:18:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d128be5c6f7400ba1bd653003af263d : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c8141bc64684733bad2001215f63888 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/743ad4c885604b8284fc2bcec49fa4b7 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb1ca45ab244a88b0d9dbadd6ea5782 : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":2},{"id":"29ea98c39106b219a9377cba5f633506280d93f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:18:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"37ee9fc555b3f5021dc20d7dbe69485c4339b2c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:23:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef496fa16c2447ea5cdf875cd1e3e31 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8a71c83adb30293f3aea4e55fcb686c3d9aa6c4","revisions":{"f8a7dd0c3742f0daa787eb9ee1ac2b5014b020cc":{"kind":"REWORK","_number":1,"created":"2020-06-16 10:48:28.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/67/735867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/67/735867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/735867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/735867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/735867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/67/735867/1"}}},"commit":{"parents":[{"commit":"d2453d77b6055da9cea71dba6353b0d55df69771","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/d2453d77b6055da9cea71dba6353b0d55df69771"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-06-16 10:47:30.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-06-16 10:48:26.000000000","tz":120},"subject":"Add a /healthcheck by default","message":"Add a /healthcheck by default\n\nThis patch adds a /healthcheck URL by default, using oslo.middleware,\nso that operators can use it to configure haproxy / monitoring.\n\nChange-Id: Idf9a184f969e5f77de0c2f2a2b99c2931f5647e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/f8a7dd0c3742f0daa787eb9ee1ac2b5014b020cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/f8a7dd0c3742f0daa787eb9ee1ac2b5014b020cc"}]},"branch":"refs/heads/master"},"e8a71c83adb30293f3aea4e55fcb686c3d9aa6c4":{"kind":"REWORK","_number":2,"created":"2020-07-02 09:27:25.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/67/735867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/67/735867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/735867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/735867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/735867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/67/735867/2"}}},"commit":{"parents":[{"commit":"d2453d77b6055da9cea71dba6353b0d55df69771","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/d2453d77b6055da9cea71dba6353b0d55df69771"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-06-16 10:47:30.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-07-02 09:26:09.000000000","tz":120},"subject":"Add a /healthcheck by default","message":"Add a /healthcheck by default\n\nThis patch adds a /healthcheck URL by default, using oslo.middleware,\nso that operators can use it to configure haproxy / monitoring.\n\nChange-Id: Idf9a184f969e5f77de0c2f2a2b99c2931f5647e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/e8a71c83adb30293f3aea4e55fcb686c3d9aa6c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/e8a71c83adb30293f3aea4e55fcb686c3d9aa6c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
