)]}'
{"id":"openstack%2Fpanko~752048","triplet_id":"openstack%2Fpanko~stable%2Frocky~Icf83ffe089301b3782273923f18efd4d209131c2","project":"openstack/panko","branch":"stable/rocky","topic":"expire-stable/ussuri-stable/train-stable/stein-stable/rocky","hashtags":[],"change_id":"Icf83ffe089301b3782273923f18efd4d209131c2","subject":"Add max_count argument for clear_expired_data call","status":"ABANDONED","created":"2020-09-15 13:32:06.000000000","updated":"2020-09-23 09:31:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6d0cbdfe1a8192a275a9adef002073fdd2309c9f","_number":752048,"virtual_id_number":752048,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"date":"2020-09-21 12:30:42.000000000","_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-15 16:59:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2020-09-15 13:32:06.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-15 13:32:06.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-09-15 16:59:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"342c5483963ec031a020212779fb1722ed606160","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-15 13:32:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"bcfd3999efc79a6dc1df2e7adfe89ecec70bf5db","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-15 13:32:17.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit af67d2cd0f54af17e5a247645664a87a7891a218","accounts_in_message":[],"_revision_number":1},{"id":"e0d29ec77f2b68eb98a715d799895c0bae005c9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 16:59:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a74529c5454868a21429cd566583e9 : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4117628fb795443b85808c05024b5a0f : FAILURE in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe916c4a681b44c2a9bc5f61b2ef329c : POST_FAILURE in 4m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7e7f24ba0dbb448e8a8bacfdb957de35 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3423aa14b57b40e99eed231510ffd8f7 : SUCCESS in 5m 15s\n- panko-tox-py27-mongodb https://zuul.opendev.org/t/openstack/build/34969249e9b94650936ddc6814e0134b : POST_FAILURE in 5m 56s\n- panko-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/1b272a5d30884cd8be09433b1187c9ad : POST_FAILURE in 5m 33s\n- panko-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/f268eeea649a40c8bfe818e815e3ed59 : FAILURE in 4m 17s\n- panko-tox-py35-mongodb https://zuul.opendev.org/t/openstack/build/cad6871d59584ccc8781f63e65e6aa13 : SUCCESS in 6m 32s\n- panko-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/ecaf18508d1443c1b2b67bff9d151e6e : SUCCESS in 7m 40s\n- panko-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/b2cd291fb3624cf4a556e91895a03f4d : SUCCESS in 7m 05s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c825fdd9aca7486c8f444aef3c12a023 : FAILURE in 43m 35s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/23dc016973c74a8897553a9e6b5ae0a7 : FAILURE in 1h 46m 12s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5a1fa74b3b734965974a4bc9b3261847 : TIMED_OUT in 3h 20m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc482915bce509c0cdce5b3db1db87b693c8fdcc","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-21 12:30:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b3849d3ba9072ef54f3331ce3845d06718006fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 15:41:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d2edfa8867d414aa49d5bfdf21c2d9f : SUCCESS in 7m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aca4aa960f044d1d947cac1b6da04698 : FAILURE in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c99b65fc885648539a0709340fef1e19 : SUCCESS in 5m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fd95b106daba42b88796eb50f8974ad2 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c365a191be24ef3ae55ac12b396ca2a : SUCCESS in 5m 56s\n- panko-tox-py27-mongodb https://zuul.opendev.org/t/openstack/build/705bb63f5017432db236ced7e192113b : FAILURE in 4m 31s\n- panko-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/dbb5fc312c524b669a3fc6403ab93bb8 : FAILURE in 6m 38s\n- panko-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/2a35bc5c5e114cd9b3a3f7a396ca3372 : FAILURE in 3m 52s\n- panko-tox-py35-mongodb https://zuul.opendev.org/t/openstack/build/fecdef364bbd4f7b9330725c762640ec : SUCCESS in 5m 35s\n- panko-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/bfe1c7b8efb944c89006518539babea9 : SUCCESS in 5m 50s\n- panko-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/882dba888a3d49ed9e1c203bb52c0851 : SUCCESS in 6m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/abea1618b2d342af971461d0b1d898ff : FAILURE in 47m 04s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/dec0b21ea3374d54a3c28d4eaca8a0f4 : SUCCESS in 2h 53m 48s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/813f9baca6d546129fd7238c8a30ab7c : SUCCESS in 2h 50m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c214acf4fb05c48aeca04a74a254d5a65938a343","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-23 09:31:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be7f284fe293c599ab605cbfac5a475cc90ba3de","revisions":{"be7f284fe293c599ab605cbfac5a475cc90ba3de":{"kind":"REWORK","_number":1,"created":"2020-09-15 13:32:06.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/48/752048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/48/752048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/48/752048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/48/752048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/48/752048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/48/752048/1"}}},"commit":{"parents":[{"commit":"1c4399d58e3946d3d0364a9a592352e05423a2a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/1c4399d58e3946d3d0364a9a592352e05423a2a6"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-06-03 11:11:09.000000000","tz":180},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2020-09-15 13:32:06.000000000","tz":0},"subject":"Add max_count argument for clear_expired_data call","message":"Add max_count argument for clear_expired_data call\n\nWhen thare\u0027re a lot of events in the database panko-expire\nconsumes a lot of memory to delete all records with a single\ncall. To avoid this behaviour a new config option\n`events_delete_batch_size` was introduced.\n\nChange-Id: Icf83ffe089301b3782273923f18efd4d209131c2\n(cherry picked from commit ec84ed8aa7c75d83b50ddac3edb4025462a61084)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/be7f284fe293c599ab605cbfac5a475cc90ba3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/be7f284fe293c599ab605cbfac5a475cc90ba3de"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
