)]}'
{"id":"openstack%2Fpanko~763388","triplet_id":"openstack%2Fpanko~stable%2Ftrain~Ib3e2f430914f5300cc9be053b6c3867d7eefd93b","project":"openstack/panko","branch":"stable/train","topic":"ci-pass","hashtags":[],"change_id":"Ib3e2f430914f5300cc9be053b6c3867d7eefd93b","subject":"Make telemetry-dsvm-integration non-voting","status":"MERGED","created":"2020-11-19 14:52:01.000000000","updated":"2020-11-23 09:15:59.000000000","submitted":"2020-11-23 09:05:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763388-ci-pass","meta_rev_id":"5542b6fd28c8a84639e415ca8eab405c42ca8b34","_number":763388,"virtual_id_number":763388,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"date":"2020-11-20 13:48:59.000000000","_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-23 09:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2020-11-23 08:12:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2020-11-23 08:12:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-20 14:38:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-23 06:52:08.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2020-11-23 08:12:52.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"1af7e51e57fb9dc510b1bb2b4278c8147ba5693b","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-11-19 14:52:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20be50246802c5946d82abee527c65e6a9aa215e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:52:21.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/panko on branch stable/train.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/panko/.zuul.yaml@stable/train\", line 10, column 19","accounts_in_message":[],"_revision_number":1},{"id":"3a5d37dc7d26de6b99d052398d3087be3e21d36b","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-11-19 22:27:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2d6e1004d6a447b4dc5487dff3b4f57951b4d7f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 23:14:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7859dafa22bb4ea08445677ffa0993c8 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef1955603fa74fbf871aebe81f1e3135 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cacdc41b610e466f8cf533e86ec7e1be : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/209b4e66067f47be84e0b89daa76aab6 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/699369d886f341db99f4d1feed6cedb7 : SUCCESS in 7m 27s\n- telemetry-dsvm-integration-train https://zuul.opendev.org/t/openstack/build/cdf1166dd122486e84b6acd11585d35c : POST_FAILURE in 39m 19s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c42b453607b5430b8216cc0d2b0c795a : POST_FAILURE in 40m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9d5e717891c86e9965e1c81854c2b66823aac02e","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-11-20 13:48:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"70651180f3c7598dac41ee4f11f838e1dff6c63a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 14:38:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f5f9d0abdf1428c98ffe8cfd59bbb18 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc7356f1cfee4d1288ef41ef7366e9d2 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a9dd17eadff4f4282f350680655b298 : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b98e5b7b98384a8a839145a466dae363 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c6a4f57abc44368de354303e304a80 : SUCCESS in 6m 36s\n- telemetry-dsvm-integration-train https://zuul.opendev.org/t/openstack/build/10104a0e44e74ccfa8bb77d6c6bb0284 : SUCCESS in 43m 10s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/30c39152fe7e44139b415f490f6d1dd3 : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0667cb7759f5707e0462291c78453b48e0dd6903","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-11-23 06:52:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"922ea5f72d554593ff70aaa5e986c93f1193ad5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 07:45:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d06d34c7a72452a8234fe5748b7b1df : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f431dddc2fce4a3e91a3d6cd8a3429dd : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/deec58bcae2644b7a94bf727ed3f7f43 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7287bc203694cd9849977a1ad428066 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6db3d02c35e40299de85e32945348b5 : SUCCESS in 8m 11s\n- telemetry-dsvm-integration-train https://zuul.opendev.org/t/openstack/build/f334357f2bd24b02a4b456e9ccc99f9e : SUCCESS in 45m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b43490da8c77404eb3f4f97026d74c7b : SUCCESS in 46m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b29a0fd0061116d55d0dad2309dc6b72bc4549f7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-11-23 08:12:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c5a598f3678c4364582155a4cf8127a5df78098b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 08:13:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b94a3feee94d072618b82d3df9b78addd2a0f12e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 09:05:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef87bf8d47e460e8326c901fe925a18 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7f4a6fd20e244e39e25a04dac7ff486 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ce3bc8275204cd18806984f5563d9bb : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8e9153f73824832b116bf816e2b38ee : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427645a14a8b4f1a9d02a537adcfc101 : SUCCESS in 6m 45s\n- telemetry-dsvm-integration-train https://zuul.opendev.org/t/openstack/build/47ff3127a04345a8af80184403ee59ad : SUCCESS in 46m 32s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ff447a3841044e4b881f43e4354b65f5 : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6151e4449949c0b85cb12a30569703bf1c0ccdb0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 09:05:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5542b6fd28c8a84639e415ca8eab405c42ca8b34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 09:15:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa0d54d918b44a09c9e813fbe4334f2 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c19b58c539ea6d34c33b05911bbbd2a8af7bb2c7","revisions":{"33b082612644f041146f2799df96897c041732db":{"kind":"REWORK","_number":1,"created":"2020-11-19 14:52:01.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/88/763388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/88/763388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/88/763388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/88/763388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/88/763388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/88/763388/1"}}},"commit":{"parents":[{"commit":"9b551e7a4501a5d5109adc55cc9cf8c314444d0c","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/9b551e7a4501a5d5109adc55cc9cf8c314444d0c"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-11-19 14:03:07.000000000","tz":60},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-11-19 14:51:59.000000000","tz":60},"subject":"Make telemetry-dsvm-integration non-voting","message":"Make telemetry-dsvm-integration non-voting\n\nThe job currently fails in train for irrelevant reasons.\n\nChange-Id: Ib3e2f430914f5300cc9be053b6c3867d7eefd93b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/33b082612644f041146f2799df96897c041732db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/33b082612644f041146f2799df96897c041732db"}]},"branch":"refs/heads/stable/train"},"c19b58c539ea6d34c33b05911bbbd2a8af7bb2c7":{"kind":"REWORK","_number":2,"created":"2020-11-19 22:27:34.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/88/763388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/88/763388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/88/763388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/88/763388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/88/763388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/88/763388/2"}}},"commit":{"parents":[{"commit":"9b551e7a4501a5d5109adc55cc9cf8c314444d0c","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/9b551e7a4501a5d5109adc55cc9cf8c314444d0c"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-11-19 14:03:07.000000000","tz":60},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-11-19 22:27:24.000000000","tz":60},"subject":"Make telemetry-dsvm-integration non-voting","message":"Make telemetry-dsvm-integration non-voting\n\nThe job currently fails in train for irrelevant reasons.\n\nChange-Id: Ib3e2f430914f5300cc9be053b6c3867d7eefd93b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/c19b58c539ea6d34c33b05911bbbd2a8af7bb2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/c19b58c539ea6d34c33b05911bbbd2a8af7bb2c7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
