)]}'
{"id":"openstack%2Fpanko~763941","triplet_id":"openstack%2Fpanko~master~I6f120922c5101987a0eea08bf7e0b2b05332fddd","project":"openstack/panko","branch":"master","topic":"devstack","hashtags":[],"change_id":"I6f120922c5101987a0eea08bf7e0b2b05332fddd","subject":"Remove deprecated tail_log function","status":"MERGED","created":"2020-11-24 08:57:26.000000000","updated":"2020-11-27 07:56:04.000000000","submitted":"2020-11-27 07:53:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763941-devstack","meta_rev_id":"b232d02cf8f7ec4ec5dd814e47b8e6a33c3b51ad","_number":763941,"virtual_id_number":763941,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 07:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2020-11-27 07:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2020-11-27 07:30:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 09:21:59.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-11-24 09:48:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-27 04:49:08.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"CC"},{"updated":"2020-11-27 04:49:08.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"state":"REVIEWER"},{"updated":"2020-11-27 05:00:06.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"2c27871674c106205244e27c5299c97363a5b1d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-11-24 08:57:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f465875ec47e645101e087d5d7fec6e6f8661f3b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-11-24 09:02:00.000000000","message":"Topic set to devstack","accounts_in_message":[],"_revision_number":1},{"id":"bd4e53484c1ba74ec86158ddef96c0fa03e6415a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-11-24 09:21:59.000000000","message":"Patch Set 1: Code-Review+2\n\nThank you, looks good to me","accounts_in_message":[],"_revision_number":1},{"id":"33f2a56fefbd4e0bfe52cf3719b7d147829d0a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 09:48:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b67e552f5b554e9ea591be0d10e364d0 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89ff6f62f0934ad9ba60b3983f9d3518 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f3da127f2da40919dd83cbb60f49828 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b3be1b5d6574b1b9b4d04797cdc93c8 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c6bb444c56345a183321cb006460d5f : SUCCESS in 6m 37s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/694a7befdf9d44aaae3e0fee01dbc2fd : FAILURE in 44m 31s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/55e8ad49a2c1400a8471e9a7f0ee964f : FAILURE in 44m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a0ab5910967dc655cce90385c41990f29e87a66c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-11-24 10:00:02.000000000","message":"Patch Set 1:\n\nthere seems to be a circular dependency between aodh and panko\nmaybe make it no-voteable merge panko and aodh and then make it voteable again ?","accounts_in_message":[],"_revision_number":1},{"id":"a271664b49ce8699fae361616ec749a79c2791d9","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-11-27 07:03:10.000000000","message":"Patch Set 1:\n\nyes, temporarily removing votable from the job totally makes sense to me.","accounts_in_message":[],"_revision_number":1},{"id":"1fc1f8c1c3e2917ee54f1000f19f4bf741b19ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-11-27 07:09:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"10fd3ea11368cce185927cb1e61e96708466f0e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 07:21:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ee381d7dc0441ca7e6a6a8ec135997 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fab24d601fd04218880e373533177e59 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4671e9ea5104d3bb1a47ed6e6462ee2 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/623d899c5a6f4faa84697f0299b17a59 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0372ca2fc1974b979d1796caf2de76ee : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d76d72d54e714d9b01efd86ff7f1c29bf9d57b9b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-11-27 07:30:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9be17fb714d42c1e4c77464b4cbf0ffbac809f00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 07:30:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"933b8d9a606a5ee1098c19ae22f8ad10b5e650c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 07:53:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9dc6be62b7a4d8894e6b332122f2a22 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd79c830fa44405598de90bcc0a8c08c : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7af84f511ae34d019ff14c3381f72ffc : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acfd5b0d5f4b41bb8f5c76e0169375c6 : SUCCESS in 13m 49s","accounts_in_message":[],"_revision_number":2},{"id":"31bfd65757530972ae7d44c156c18c9356a9eb32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 07:53:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b232d02cf8f7ec4ec5dd814e47b8e6a33c3b51ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 07:56:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0eb10e56f8747a5b4f05ae569566c0a : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a7684460bff5009359c2c814643ce0507b545d6","revisions":{"318aaf04ad8d671dfd80aab84a76183a66d81dee":{"kind":"REWORK","_number":1,"created":"2020-11-24 08:57:26.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/41/763941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/41/763941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/41/763941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/41/763941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/41/763941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/41/763941/1"}}},"commit":{"parents":[{"commit":"5562d835dde7cc32a67d63211e770d6daa1057ad","subject":"Merge \"Remove install unnecessary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/5562d835dde7cc32a67d63211e770d6daa1057ad"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-11-24 08:56:10.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-11-24 08:57:21.000000000","tz":120},"subject":"Remove deprecated tail_log function","message":"Remove deprecated tail_log function\n\nThis function has been deprecated for a long time, let\u0027s finally\nremove it. It is only generating a warning anyway.\n\nChange-Id: I6f120922c5101987a0eea08bf7e0b2b05332fddd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/318aaf04ad8d671dfd80aab84a76183a66d81dee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/318aaf04ad8d671dfd80aab84a76183a66d81dee"}]},"branch":"refs/heads/master"},"0a7684460bff5009359c2c814643ce0507b545d6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-27 07:09:11.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/41/763941/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/41/763941/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/41/763941/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/41/763941/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/41/763941/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/41/763941/2"}}},"commit":{"parents":[{"commit":"5562d835dde7cc32a67d63211e770d6daa1057ad","subject":"Merge \"Remove install unnecessary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/5562d835dde7cc32a67d63211e770d6daa1057ad"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-11-24 08:56:10.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-11-27 07:08:53.000000000","tz":0},"subject":"Remove deprecated tail_log function","message":"Remove deprecated tail_log function\n\nThis function has been deprecated for a long time, let\u0027s finally\nremove it. It is only generating a warning anyway.\n\nDepends-On: https://review.opendev.org/c/openstack/panko/+/764418\nChange-Id: I6f120922c5101987a0eea08bf7e0b2b05332fddd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/0a7684460bff5009359c2c814643ce0507b545d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/0a7684460bff5009359c2c814643ce0507b545d6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
