)]}'
{"id":"openstack%2Fpanko~764946","triplet_id":"openstack%2Fpanko~master~I222155c8c6b757fd80b30173fd300e2a7240c107","project":"openstack/panko","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I222155c8c6b757fd80b30173fd300e2a7240c107","subject":"Update requirements for secure RBAC work","status":"MERGED","created":"2020-12-01 14:43:10.000000000","updated":"2021-01-18 17:37:51.000000000","submitted":"2021-01-18 17:35:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764946-secure-rbac","meta_rev_id":"584a1b2c89399a6a86248744df00a37a79137b47","_number":764946,"virtual_id_number":764946,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 17:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2020-12-03 09:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-12-11 08:32:50.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2020-12-11 08:32:50.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 15:24:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 09:55:04.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-12-11 08:32:50.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"b4afc97f5ff8d29ae8702e913c1d6acb96be9d04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-01 14:43:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f8bc42156ab0ca2bc2421a04420543158760573","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-01 14:44:51.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"857945cb83f1da0dfe2f138e163679049045be8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:24:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6799a660ce047faad07a3bc4284133e : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e56449194e74103b588f0e47ca47763 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/596676f0544f4002a18b9350cfd7a5b7 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc8c31a952804116a3a3566af5c08dae : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640b63ffcd41489a8a08e75576236acb : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4fc48e3b5168a290484481f8d90ebca94ad5c0d1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-03 09:55:04.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"fc2d598c3cf8499be631a1ace94512a4292cd430","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-12-11 08:32:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"038d42cb0975cb5a7e04c5cf1e132a7ac8dfae80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 08:33:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef5999a8ba1861978fe9fc7e5ad5f615d0226831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 08:53:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d043af1be51417eaf645e9b54d942c6 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8d7be616d71492da0bb2bab4ab13664 : FAILURE in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66a45066718b42f6bb7e5962abad70a1 : FAILURE in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5656e726c9405eb05180c272a5aca2 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3c2a4f1a5e75b45259e01280a5eb3d82724c25ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 08:54:12.000000000","message":"Patch Set 1:\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d043af1be51417eaf645e9b54d942c6 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8d7be616d71492da0bb2bab4ab13664 : FAILURE in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66a45066718b42f6bb7e5962abad70a1 : FAILURE in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5656e726c9405eb05180c272a5aca2 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"aafed7d5fb2a768d528c0c64a4defff36a1dec06","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-11 09:16:24.000000000","message":"Patch Set 1:\n\nfails with \nERROR: No matching distribution found for fixtures\u003c2.0,\u003e\u003d1.3.1","accounts_in_message":[],"_revision_number":1},{"id":"0748e46830157aa9158168d48abbe00b85133f65","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-12 08:54:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0c8eb9b10c56da6df8647af4ad07343fe10d899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 09:15:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05dfe2d1acb4f55b40d3ccf2f87e67f : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09f5996b4ded4766a6b68ca37a46b1ba : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/870adf7272974acdb9296acbe8a38455 : FAILURE in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1c553365b8a4a52a45b2784af8ab011 : FAILURE in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597d6fd1c0064edea6bb1688ee8e72d0 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"075a49aa788080588bfb82f9b5857543c6ea05c3","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-13 22:02:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3c800670f42d8de82ceb078d6f5a678eceb626fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 22:21:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f4a35d74084aa28594f6fcc2d58f6a : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aaef073ce9940739e3b44a161179f2a : FAILURE in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df5124e1676c4bc88e0b04ec5f84a481 : FAILURE in 12m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b3819b249174eab962742a00d027219 : FAILURE in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efe5d520fdd43c4bacbd2443c9d4d76 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":1},{"id":"35378d57e800fac6917a2a4e5779942969109f3d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-18 16:45:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fccf5ce47023ffd73fde36be9f49f648025a95f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:22:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94336496beee4be7ac9c39c5b3b997ed : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab24913a80aa494ea3d76c00ca0438bf : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e361c413a4f5412880f96db12054544a : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4a7a19149a142c98c3c86bd2c94ad5a : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feebc5fecd44426f96cfe07543a7802e : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c4a8e6bd19e6657c51a4509b443952db0f37617a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:22:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28d13a68fb16c98da4494a28876efab1fa17245a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:35:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56766d2f0f514805a6963e5868fe0cfe : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14cdcaa712c44e59b5cc361812d37c99 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06f11740d52c420e8584677ce9d4991c : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96047b19662b4555be8cec68c1eecb7c : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b9924a6cf08a0414e28051ebb3a01fee0d32c730","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:35:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"584a1b2c89399a6a86248744df00a37a79137b47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:37:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2d799369eb414199d94b9943c1767e : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6edbdfa296372a09c3dee14de165e09d630cb288","revisions":{"6edbdfa296372a09c3dee14de165e09d630cb288":{"kind":"REWORK","_number":1,"created":"2020-12-01 14:43:10.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/46/764946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/46/764946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/46/764946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/46/764946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/46/764946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/46/764946/1"}}},"commit":{"parents":[{"commit":"36b909a03dbacdf84482051c35416e63f3723d1e","subject":"Merge \"Remove deprecated tail_log function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/36b909a03dbacdf84482051c35416e63f3723d1e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-01 14:27:01.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-01 14:27:01.000000000","tz":0},"subject":"Update requirements for secure RBAC work","message":"Update requirements for secure RBAC work\n\nThis commit updates the versions of oslo.policy, keystonemiddleware, and\noslo.log to new versions that aid in implementing secure RBAC through\ndefault roles and scope checking.\n\nThis commit also includes a dependency on oslo.context, which provides\npythonic context objects based on request environments. This is useful\nin implementing authorization decisions because we can just pass context\nobjects to oslo.policy\u0027s Enforcer() object instead of building a\ndictionary of authorization information manually.\n\nThese dependencies will be used in a subsequent patch set to update\nPanko\u0027s default policies.\n\nChange-Id: I222155c8c6b757fd80b30173fd300e2a7240c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/6edbdfa296372a09c3dee14de165e09d630cb288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/6edbdfa296372a09c3dee14de165e09d630cb288"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
