)]}'
{"id":"openstack%2Fpanko~764947","triplet_id":"openstack%2Fpanko~master~I12074fe328db71895bbdfda3c9e7c56cc6b6f40d","project":"openstack/panko","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I12074fe328db71895bbdfda3c9e7c56cc6b6f40d","subject":"Add a system-reader check string to base.py","status":"MERGED","created":"2020-12-01 14:43:10.000000000","updated":"2021-01-18 20:01:29.000000000","submitted":"2021-01-18 19:58:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764947-secure-rbac","meta_rev_id":"63f1cb949b6cd787360f03e01954c2eeed780eda","_number":764947,"virtual_id_number":764947,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 19:58:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2020-12-03 09:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-12-11 08:34:58.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2020-12-11 08:34:58.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 15:38:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 09:56:36.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-12-11 08:34:58.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"2cdfc25d5b33455763b00c7f30f8eaf986b10d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-01 14:43:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"611278e1334bb57bdc845ce6a725c7abd3fb915e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-01 14:45:12.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"f8a9ce7b5aca52afc0f93bfa7677df825ec96409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:38:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c43600f69eb45ee98efb2fcbffe451e : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87595a495abf4504a8eff236599b1be1 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/810552c8e52f46b7b054feb1efb68497 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58e565c156794de5a86d4d93a51574e6 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d4e1dbe26a4b34904c337edc0650e1 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"77fee7b4a216a87699791bae3e68632860391f19","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-03 09:56:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f3aae362b2fd1ea660cd2468dcfff244f8e070b8","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-12-11 08:34:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1c1000a2a67851f709f3d39f730f5e827250d55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 08:35:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4fd4aac5325255317de3c1c09e3a7c013558f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 08:59:56.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"edcc7769392a9d57f72a461496121d9034dbeab5","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-12 08:54:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0aac26db459ea0df9efac6e5755e59c21b20b015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:23:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4315d021fd6423a857128df822a9048 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/663bc8a593f84dd481c92f9db527e24f : FAILURE in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40beb53262364329b28ac9abc01f0345 : FAILURE in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d232d3c56cac46ae8fa725723e288e00 : FAILURE in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db2042051734430ae347ebeec663c94 : SUCCESS in 7m 47s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/58efcdf5070149e7b6336cb2e79cc032 : SUCCESS in 47m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/dbb0caa13e414b07b23936fc93cd7be1 : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"47da7c09c1b5919b36b41e022f312dbc961ef9d6","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-18 17:45:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3567309378964e98e9723edfacd0e9016ec0b1e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:57:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3292a300e4ff4cd5b98240a32be2c1c1 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/066736e802bf4f2b8f0011e498b655ec : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0505e216c774f809861bfab8581a8fa : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72ab502039374cab80f60c52fd53e851 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3200420ba44860817088f3bcd8a341 : SUCCESS in 6m 49s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/afd3e80d3de8499c84e8a6fc5605d0c1 : SUCCESS in 53m 53s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d445f757e1de4d2e961f4ae284cc7499 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a50379da7e344ab36b506e224574aa09cf756447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:58:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"816901a1c70f78a3cc07f2367667e6fcb24f12c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 19:58:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/092319e4f9034c7b9840efec6ff83883 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1decbbb9cfb44b3916fc8e6ac0c3496 : SUCCESS in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a26e066a63940828945a05a5473d1a7 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef3a4ed2c32340de8d106effb5ec2350 : SUCCESS in 6m 39s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a71cefd9dfaa447ebb36bc1e2b0daea8 : SUCCESS in 48m 47s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c8b0d6631a43493ebb4fe520a6b2f173 : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3fac8bd787b430d03278298c835ffe4d964900ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 19:58:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"63f1cb949b6cd787360f03e01954c2eeed780eda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 20:01:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e824ae8f184f62a4520810fc49701e : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12c41a0a73114974b8d5d15a1cc14368d68099cc","revisions":{"12c41a0a73114974b8d5d15a1cc14368d68099cc":{"kind":"REWORK","_number":1,"created":"2020-12-01 14:43:10.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/47/764947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/47/764947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/47/764947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/47/764947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/47/764947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/47/764947/1"}}},"commit":{"parents":[{"commit":"6edbdfa296372a09c3dee14de165e09d630cb288","subject":"Update requirements for secure RBAC work","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/6edbdfa296372a09c3dee14de165e09d630cb288"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-01 14:35:41.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-01 14:35:41.000000000","tz":0},"subject":"Add a system-reader check string to base.py","message":"Add a system-reader check string to base.py\n\nThis commit adds a new constant to base.py that models a common persona\nbeing implemented across OpenStack called system-reader. We can use this\npersona in future patches to update the default policies for the\nappropriate APIs.\n\nChange-Id: I12074fe328db71895bbdfda3c9e7c56cc6b6f40d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/12c41a0a73114974b8d5d15a1cc14368d68099cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/12c41a0a73114974b8d5d15a1cc14368d68099cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
