)]}'
{"id":"openstack%2Fpanko~766656","triplet_id":"openstack%2Fpanko~master~I199aa85acd2dc920a529cde52ec060f1904b869c","project":"openstack/panko","branch":"master","hashtags":[],"change_id":"I199aa85acd2dc920a529cde52ec060f1904b869c","subject":"Remove six.moves","status":"ABANDONED","created":"2020-12-11 02:56:40.000000000","updated":"2020-12-11 07:24:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70c86eaf5199e4354d57e027fa334c524604a2bc","_number":766656,"virtual_id_number":766656,"owner":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-11 03:25:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-11 03:25:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ab1ae9019e7888dcdc151341b15218c915ac17b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"date":"2020-12-11 02:56:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51c265401160cfa1366209a7ae4c8e72846f32b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 03:25:03.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f678637f71724ce7906678b1b2ad903d : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79969f6c179e49078b528bb5141bae5f : FAILURE in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f831f073d1340d09ab81555f9071bb5 : FAILURE in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff52a6f63bdf45e1b69c85c676e352d2 : FAILURE in 3m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec68478e3c0e4c9e80276f4358185dbf : SUCCESS in 6m 12s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/25bb4cb73a6a4dbdb864f4482a42cda8 : FAILURE in 27m 09s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ba009f59081a484db13da2b7b60847e3 : FAILURE in 25m 46s","accounts_in_message":[],"_revision_number":1},{"id":"70c86eaf5199e4354d57e027fa334c524604a2bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"date":"2020-12-11 07:24:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c1b8c362558e6700545f2016dcd4334d2e18503","revisions":{"2c1b8c362558e6700545f2016dcd4334d2e18503":{"kind":"REWORK","_number":1,"created":"2020-12-11 02:56:40.000000000","uploader":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"ref":"refs/changes/56/766656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/56/766656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/56/766656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/56/766656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/56/766656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/56/766656/1"}}},"commit":{"parents":[{"commit":"ac92c378b89b6dcb98193ee902b71588f50841d6","subject":"Merge \"Revert \"Temporarily switch dvsm-integration off\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ac92c378b89b6dcb98193ee902b71588f50841d6"}]}],"author":{"name":"suhaiming","email":"suhaiming@inspur.com","date":"2020-12-11 02:50:13.000000000","tz":0},"committer":{"name":"suhaiming","email":"suhaiming@inspur.com","date":"2020-12-11 02:56:20.000000000","tz":0},"subject":"Remove six.moves","message":"Remove six.moves\n\nChange-Id: I199aa85acd2dc920a529cde52ec060f1904b869c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/2c1b8c362558e6700545f2016dcd4334d2e18503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/2c1b8c362558e6700545f2016dcd4334d2e18503"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
