)]}'
{"id":"openstack%2Fpanko~766686","triplet_id":"openstack%2Fpanko~master~I083378314cd5f593e90b5b33e4921de47f32a494","project":"openstack/panko","branch":"master","hashtags":[],"change_id":"I083378314cd5f593e90b5b33e4921de47f32a494","subject":"remove six","status":"ABANDONED","created":"2020-12-11 08:31:42.000000000","updated":"2024-10-22 05:04:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a632db8f1ced08349c58094d56fa85484177a3e","_number":766686,"virtual_id_number":766686,"owner":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-11 09:14:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-11 09:14:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de78a9e3a1b5f652203b8db606f328910a3335b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"date":"2020-12-11 08:31:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0b3a884e606e7214680170e3d0a4f0c0e199fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 09:14:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1574de1a0ff84db8ad761a9800d785e3 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9007ddfed68e4f2e8908df87bb68787a : FAILURE in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ab0aba9e5f9449782abfcdf61a96b5b : FAILURE in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4987fcbdbe4948b2978ce695120a09fe : FAILURE in 3m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9b445ae074b42458f219a9c5a8df2d7 : SUCCESS in 7m 14s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/584819a1534b4e5090cb229acbc25dd8 : FAILURE in 36m 04s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b7b146a3c6664c7ea4794fc50338b09d : FAILURE in 30m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2a632db8f1ced08349c58094d56fa85484177a3e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-22 05:04:58.000000000","message":"Abandoned\n\npanko was already retired","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62aa5d457ad40789d26f7d6e967c82bda712ae07","revisions":{"62aa5d457ad40789d26f7d6e967c82bda712ae07":{"kind":"REWORK","_number":1,"created":"2020-12-11 08:31:42.000000000","uploader":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"ref":"refs/changes/86/766686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/86/766686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/86/766686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/86/766686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/86/766686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/86/766686/1"}}},"commit":{"parents":[{"commit":"ac92c378b89b6dcb98193ee902b71588f50841d6","subject":"Merge \"Revert \"Temporarily switch dvsm-integration off\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ac92c378b89b6dcb98193ee902b71588f50841d6"}]}],"author":{"name":"suhaiming","email":"suhaiming@inspur.com","date":"2020-12-11 08:31:15.000000000","tz":0},"committer":{"name":"suhaiming","email":"suhaiming@inspur.com","date":"2020-12-11 08:31:15.000000000","tz":0},"subject":"remove six","message":"remove six\n\nChange-Id: I083378314cd5f593e90b5b33e4921de47f32a494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/62aa5d457ad40789d26f7d6e967c82bda712ae07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/62aa5d457ad40789d26f7d6e967c82bda712ae07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
