)]}'
{"id":"openstack%2Fpanko~800167","triplet_id":"openstack%2Fpanko~master~I5acab95813ac3d0604f73c377bf111f423a008bc","project":"openstack/panko","branch":"master","topic":"really_remove","hashtags":[],"change_id":"I5acab95813ac3d0604f73c377bf111f423a008bc","subject":"Remove leftovers of retirement","status":"MERGED","created":"2021-07-09 08:58:07.000000000","updated":"2021-07-22 15:01:17.000000000","submitted":"2021-07-22 15:01:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800167-really_remove","meta_rev_id":"bd5e3adce29bc38546392d4973d93ceb06aedb90","_number":800167,"virtual_id_number":800167,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 15:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2021-07-22 15:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2021-07-22 15:01:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 09:01:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-09 13:22:49.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"cad749cbebb0202b1421b2836273f1a765103d59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 08:58:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cdce2312537839fab3687418e469fa3f8b1df8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 09:01:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/54f0be49bd9c44b882483d61a380f094 : FAILURE in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9977615f4e21ff05662dcc24f432cac177226892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 13:21:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d86847a484791bfbfa04cc6b65172496f4c996d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 13:22:49.000000000","message":"Patch Set 1:\n\nThis is basically the same patch as Takashi submitted. I have incorporated the feedback on his patch here.","accounts_in_message":[],"_revision_number":1},{"id":"4297f17be9dcc165f615bff57068866a7384d106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:25:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2a69a9bcfaee401e8afa792715378215 : FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"25023b006ddbe2990e706aa1942c8e637004f92c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-22 14:56:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d59204de7a6bf9c726b5c1d35c405280b683c060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 14:57:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/44daaa004c414e448ec03032b232c23b : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"818916714315566c2ab70274f6d36a970a99c772","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-22 15:01:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1c0355e71caee8c48d739697168c06682ab9ac38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:01:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"269fbaf27441ae9ea83528d288b1b96eaf8271b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:01:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/5600a63b6e3c486fa07524da6714e378 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"bd5e3adce29bc38546392d4973d93ceb06aedb90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 15:01:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"92be85c4ca1ce3f5cf3f170e8bccd4919d3fa37b","revisions":{"d1d5d917b02abfa42d23a2bafae346c4c3861f7e":{"kind":"REWORK","_number":1,"created":"2021-07-09 08:58:07.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/67/800167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/67/800167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/800167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/800167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/800167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/67/800167/1"}}},"commit":{"parents":[{"commit":"ca45bbdca049c6544cd9ece0d564ae7f20a1cd78","subject":"Retire panko, the code is no longer maintained.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ca45bbdca049c6544cd9ece0d564ae7f20a1cd78"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-09 08:57:45.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-09 08:57:45.000000000","tz":120},"subject":"Remove leftovers of retirement","message":"Remove leftovers of retirement\n\nChange-Id: I5acab95813ac3d0604f73c377bf111f423a008bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/d1d5d917b02abfa42d23a2bafae346c4c3861f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/d1d5d917b02abfa42d23a2bafae346c4c3861f7e"}]},"branch":"refs/heads/master"},"92be85c4ca1ce3f5cf3f170e8bccd4919d3fa37b":{"kind":"REWORK","_number":2,"created":"2021-07-09 13:21:58.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/67/800167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/67/800167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/800167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/800167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/67/800167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/67/800167/2"}}},"commit":{"parents":[{"commit":"ca45bbdca049c6544cd9ece0d564ae7f20a1cd78","subject":"Retire panko, the code is no longer maintained.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/ca45bbdca049c6544cd9ece0d564ae7f20a1cd78"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-09 08:57:45.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-09 13:21:47.000000000","tz":120},"subject":"Remove leftovers of retirement","message":"Remove leftovers of retirement\n\nChange-Id: I5acab95813ac3d0604f73c377bf111f423a008bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/92be85c4ca1ce3f5cf3f170e8bccd4919d3fa37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/92be85c4ca1ce3f5cf3f170e8bccd4919d3fa37b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
