)]}'
{"id":"openstack%2Fpanko~809555","triplet_id":"openstack%2Fpanko~stable%2Ftrain~I959ca9795e753051e03be1f9000e50b13517ee72","project":"openstack/panko","branch":"stable/train","topic":"remove-c7-tripleo-jobs-stable/train","hashtags":[],"change_id":"I959ca9795e753051e03be1f9000e50b13517ee72","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-17 09:40:22.000000000","updated":"2023-12-21 14:16:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809620,"cherry_pick_of_patch_set":1,"meta_rev_id":"0ed940eabe29d52c77ffbc9665efc84a505f2849","_number":809555,"virtual_id_number":809555,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-17 10:36:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 09:40:41.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-17 09:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc5a4eafd801c41f281b0928ff9216c349c4a26f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:40:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"41f174b7dcc952777d7252628a0003745f32ee47","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-09-17 09:40:41.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/panko on branch stable/train.  The error was:\n\n  while scanning a simple key\n  in \"openstack/panko/.zuul.yaml@stable/train\", line 9, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/panko/.zuul.yaml@stable/train\", line 12, column 9","accounts_in_message":[],"_revision_number":1},{"id":"84f8d99b299496deb9b2fe224b6e37eaabca4f92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 09:40:49.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/panko on branch stable/train.  The error was:\n\n  while scanning a simple key\n  in \"openstack/panko/.zuul.yaml@stable/train\", line 9, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/panko/.zuul.yaml@stable/train\", line 12, column 9","accounts_in_message":[],"_revision_number":1},{"id":"26bd4d060eb7b6d30ab3917034baae4db61a75ba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:41:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0f36b417149bec6a860be531363b49feac2c1152","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:41:38.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"ce5542cbf97ec974ae26410faec83d47a388e01b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:36:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce437fd9b3b74bc992ef88dbce79e28c : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a674e89637ed44d0b732c9455284eb21 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f708b46294f473cb79b7b4f6b86dd9a : FAILURE in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f07b11a4c264332bc532ec9df736e91 : FAILURE in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42423dc2e22a4391863102d202be8bf2 : SUCCESS in 7m 02s\n- telemetry-dsvm-integration-train https://zuul.opendev.org/t/openstack/build/52f5cdc12c7544c3be7ba08b947324a5 : SUCCESS in 43m 13s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/77229ef602f04f51bdb485f637d598a1 : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":2},{"id":"84ab2b50184f58eac24a5707908bd2da0d7e4cc0","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-20 06:48:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8ac93b37d57cdafcae5e2083851c2ff457bd1673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:23:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1a6170b011f4cf4a2c71bfbaffe703b : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4892bd9b90f64d20b3c64b7da83d9122 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/927f69fcf52147c6a5b5dc5a9edc9b1e : FAILURE in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65cda26850414840bc387602c1f85425 : FAILURE in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2a184c65f7482d94bd0c2876edef87 : SUCCESS in 10m 24s\n- telemetry-dsvm-integration-train https://zuul.opendev.org/t/openstack/build/222f2468039c453eb6bbaff24d73ba42 : SUCCESS in 37m 12s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/86d6cae37f084cfb8b39088efa5a3d2c : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0ed940eabe29d52c77ffbc9665efc84a505f2849","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-12-21 14:16:49.000000000","message":"Abandoned\n\nTrain is about to transition to End of Life. Open patches needs to be abandoned before branch deletion.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f88e9cd64343c3417703e51908c28459923ba20a","revisions":{"16c65b66e830b961738d3bb3f12f6ac1948804c1":{"kind":"REWORK","_number":1,"created":"2021-09-17 09:40:22.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/55/809555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/55/809555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/55/809555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/55/809555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/55/809555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/55/809555/1"}}},"commit":{"parents":[{"commit":"d2ba20f241ec0fdc8f31462dbe3646117e91f736","subject":"Add max_count argument for clear_expired_data call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/d2ba20f241ec0fdc8f31462dbe3646117e91f736"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:37:47.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:40:22.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this project is no longer maintained.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I959ca9795e753051e03be1f9000e50b13517ee72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/16c65b66e830b961738d3bb3f12f6ac1948804c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/16c65b66e830b961738d3bb3f12f6ac1948804c1"}]},"branch":"refs/heads/stable/train"},"f88e9cd64343c3417703e51908c28459923ba20a":{"kind":"REWORK","_number":2,"created":"2021-09-17 09:41:30.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/55/809555/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/55/809555/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/55/809555/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/55/809555/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/55/809555/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/55/809555/2"}}},"commit":{"parents":[{"commit":"d2ba20f241ec0fdc8f31462dbe3646117e91f736","subject":"Add max_count argument for clear_expired_data call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/d2ba20f241ec0fdc8f31462dbe3646117e91f736"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:37:47.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:41:25.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this project is no longer maintained.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I959ca9795e753051e03be1f9000e50b13517ee72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/f88e9cd64343c3417703e51908c28459923ba20a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/f88e9cd64343c3417703e51908c28459923ba20a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
