)]}'
{"id":"openstack%2Fpanko~809556","triplet_id":"openstack%2Fpanko~stable%2Fstein~I959ca9795e753051e03be1f9000e50b13517ee72","project":"openstack/panko","branch":"stable/stein","topic":"remove-c7-tripleo-jobs-stable/stein","hashtags":[],"change_id":"I959ca9795e753051e03be1f9000e50b13517ee72","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-17 09:41:56.000000000","updated":"2023-09-07 14:34:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809620,"cherry_pick_of_patch_set":1,"meta_rev_id":"a15b84af849a5e1fbbd65ce68209ac6370b3c956","_number":809556,"virtual_id_number":809556,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-17 10:25:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 09:42:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-17 09:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"535a07052e70beb5a471af334cb37991d652c1df","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:41:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"5cf2e9846d679fd675bea85ba70e62443a8b175c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-09-17 09:42:15.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/panko on branch stable/stein.  The error was:\n\n  while scanning a simple key\n  in \"openstack/panko/.zuul.yaml@stable/stein\", line 11, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/panko/.zuul.yaml@stable/stein\", line 13, column 9","accounts_in_message":[],"_revision_number":1},{"id":"f809e968d6982d088d210104c66fb52554063865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 09:42:30.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/panko on branch stable/stein.  The error was:\n\n  while scanning a simple key\n  in \"openstack/panko/.zuul.yaml@stable/stein\", line 11, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/panko/.zuul.yaml@stable/stein\", line 13, column 9","accounts_in_message":[],"_revision_number":1},{"id":"e633a1359cdc54a820d9c0e8e3e55718149822da","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:42:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"507b602ab82628844c6af13853ba151d94c04ff1","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:42:41.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"7277b1d196d37a90521b874368f963ae028575ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:25:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8998d95ea4a145c7a1310431ba420778 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f29f2c2b17f478ebc3e2b9887019ddf : FAILURE in 5m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2906b13b41974642956127bed6c015d3 : TIMED_OUT in 41m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d254dc756b045fb82347fb2c461f037 : TIMED_OUT in 41m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c5cdf3f8c2943d8b30707dae6a0ad39 : TIMED_OUT in 41m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c81503ab0b047bca77087d9d608eee3 : TIMED_OUT in 31m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d5cda328fa5db5d507b84e7655724242043f3959","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-20 06:47:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e9652a4df5733e5c2c4fa3da915d270e454779a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 07:55:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5976699a743249ae8ca3a98ff032c6ed : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6985739c160243bbb5eeb7202d777562 : FAILURE in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/29674de683154514b653b972af7b2311 : TIMED_OUT in 41m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d1868f67ccb4da08f01d3c39175c4ef : TIMED_OUT in 41m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe308de9458c46479d08bda2259a9839 : TIMED_OUT in 41m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd913728e98466794a0f82f4f554464 : TIMED_OUT in 31m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a15b84af849a5e1fbbd65ce68209ac6370b3c956","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-07 14:34:05.000000000","message":"Abandoned\n\nThis project has transitioned its stable/stein branch to End of Life. All open patches need to be closed to be able to delete the branch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c16aeb3fa19dd09453979958ff14ae6fbbd404a2","revisions":{"5cb0fb5c3e9d61cbdaeda7f22b0224a6f7ea2349":{"kind":"REWORK","_number":1,"created":"2021-09-17 09:41:56.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/56/809556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/56/809556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/56/809556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/56/809556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/56/809556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/56/809556/1"}}},"commit":{"parents":[{"commit":"cc4fbd4704f89e570e36b7a5f652fadf68f3eb6f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/cc4fbd4704f89e570e36b7a5f652fadf68f3eb6f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:37:47.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:41:56.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this project is no longer maintained.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I959ca9795e753051e03be1f9000e50b13517ee72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/5cb0fb5c3e9d61cbdaeda7f22b0224a6f7ea2349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/5cb0fb5c3e9d61cbdaeda7f22b0224a6f7ea2349"}]},"branch":"refs/heads/stable/stein"},"c16aeb3fa19dd09453979958ff14ae6fbbd404a2":{"kind":"REWORK","_number":2,"created":"2021-09-17 09:42:38.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/56/809556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/56/809556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/56/809556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/56/809556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/56/809556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/56/809556/2"}}},"commit":{"parents":[{"commit":"cc4fbd4704f89e570e36b7a5f652fadf68f3eb6f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/cc4fbd4704f89e570e36b7a5f652fadf68f3eb6f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:37:47.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:42:31.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this project is no longer maintained.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I959ca9795e753051e03be1f9000e50b13517ee72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/c16aeb3fa19dd09453979958ff14ae6fbbd404a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/c16aeb3fa19dd09453979958ff14ae6fbbd404a2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
