)]}'
{"id":"openstack%2Fpanko~809557","triplet_id":"openstack%2Fpanko~stable%2Frocky~I959ca9795e753051e03be1f9000e50b13517ee72","project":"openstack/panko","branch":"stable/rocky","topic":"remove-c7-tripleo-jobs-stable/stein-stable/rocky","hashtags":[],"change_id":"I959ca9795e753051e03be1f9000e50b13517ee72","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-17 09:42:54.000000000","updated":"2023-05-23 17:59:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809556,"cherry_pick_of_patch_set":2,"meta_rev_id":"2ba4b483f03780abeaf24ff3385054be40ce5564","_number":809557,"virtual_id_number":809557,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-17 10:44:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 09:43:14.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-17 09:43:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a8d731654e0c4c4889bfbf72510fa515aa9a02d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:42:54.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"f31eb48d70fac528a79829a9522bc32f0f76724f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-09-17 09:43:14.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/panko on branch stable/rocky.  The error was:\n\n  while scanning a simple key\n  in \"openstack/panko/.zuul.yaml@stable/rocky\", line 16, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/panko/.zuul.yaml@stable/rocky\", line 17, column 9","accounts_in_message":[],"_revision_number":1},{"id":"9ef4ae583aed65d6be32bdf711df482d08a98922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 09:43:34.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/panko on branch stable/rocky.  The error was:\n\n  while scanning a simple key\n  in \"openstack/panko/.zuul.yaml@stable/rocky\", line 16, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/panko/.zuul.yaml@stable/rocky\", line 17, column 9","accounts_in_message":[],"_revision_number":1},{"id":"600f9154c0418e515786633b9e9882762ae1e600","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:43:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6f7754f76179d794b20f2ce97a1541354f9f72aa","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:43:52.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"de157d314699a4e44f13f4575e725af677f7c2d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:44:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1368b09c919b4274b6236ec80973b470 : SUCCESS in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04034a42e0ce483a8d50e547f0753ab9 : FAILURE in 5m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8e25110368544aa49c6e1df8604bfd3c : TIMED_OUT in 41m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0cd01cbf9c8640a78de6c7795d314219 : SUCCESS in 7m 15s\n- panko-tox-py27-mongodb https://zuul.opendev.org/t/openstack/build/86b6c8345645484a86835a544ab206bd : FAILURE in 5m 22s\n- panko-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/52fcd5357b024b00bb88295b56abb8da : FAILURE in 4m 25s\n- panko-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/0db33200034242d38c96ed49d6c5fa74 : FAILURE in 4m 19s\n- panko-tox-py35-mongodb https://zuul.opendev.org/t/openstack/build/072be938a2114ed49a25436ce38556b0 : TIMED_OUT in 31m 09s\n- panko-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/51044c89eb3f48509a9c59e3b191d145 : TIMED_OUT in 31m 11s\n- panko-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/a9085acb470841a2b3ed86a35a4cc60f : TIMED_OUT in 31m 11s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c1c3ad2c56d04ff0a1c88ed45c7bfff0 : FAILURE in 22m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2ba4b483f03780abeaf24ff3385054be40ce5564","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-05-23 17:59:31.000000000","message":"Abandoned\n\nrocky branch of this project is marked as End of Life. Open patches need to be abandoned in order to be able to delete the branch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bbb1544de4853acb9db5d807ffaa83ed34d65cba","revisions":{"8693a335a6643b81f745948ac57ca2d037620954":{"kind":"REWORK","_number":1,"created":"2021-09-17 09:42:54.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/57/809557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/57/809557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/57/809557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/57/809557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/57/809557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/57/809557/1"}}},"commit":{"parents":[{"commit":"1c4399d58e3946d3d0364a9a592352e05423a2a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/1c4399d58e3946d3d0364a9a592352e05423a2a6"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:37:47.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:42:54.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this project is no longer maintained.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I959ca9795e753051e03be1f9000e50b13517ee72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/8693a335a6643b81f745948ac57ca2d037620954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/8693a335a6643b81f745948ac57ca2d037620954"}]},"branch":"refs/heads/stable/rocky"},"bbb1544de4853acb9db5d807ffaa83ed34d65cba":{"kind":"REWORK","_number":2,"created":"2021-09-17 09:43:44.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/57/809557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/57/809557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/57/809557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/57/809557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/57/809557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/57/809557/2"}}},"commit":{"parents":[{"commit":"1c4399d58e3946d3d0364a9a592352e05423a2a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/1c4399d58e3946d3d0364a9a592352e05423a2a6"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:37:47.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:43:38.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this project is no longer maintained.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I959ca9795e753051e03be1f9000e50b13517ee72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/bbb1544de4853acb9db5d807ffaa83ed34d65cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/bbb1544de4853acb9db5d807ffaa83ed34d65cba"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
