)]}'
{"id":"openstack%2Fpanko~809558","triplet_id":"openstack%2Fpanko~stable%2Fqueens~I959ca9795e753051e03be1f9000e50b13517ee72","project":"openstack/panko","branch":"stable/queens","topic":"remove-c7-tripleo-jobs-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"I959ca9795e753051e03be1f9000e50b13517ee72","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-17 09:44:11.000000000","updated":"2022-12-05 18:43:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809557,"cherry_pick_of_patch_set":2,"meta_rev_id":"101efea0a0cbeab43fb595e1ab03ed3f351af093","_number":809558,"virtual_id_number":809558,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-17 10:35:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 10:35:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb139b1b5d923cf8f1fe3681e0065c52fca75a3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:44:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"c5e847cb704413122557407afb99277b0d97876a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:35:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c8774758954e029529d00e29af1fca : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbc2be9cf7634dd08d3f56aaea12abaf : FAILURE in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e87a55143ac64fb6becba917c1d444f9 : TIMED_OUT in 41m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/94621eb0d0f74ac0931aa4cd14a7af68 : SUCCESS in 5m 52s\n- panko-tox-py27-mongodb https://zuul.opendev.org/t/openstack/build/60319781fa5c465fb49874b379e32600 : FAILURE in 4m 51s\n- panko-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/cd33c06b2d544fdfb164bb1a4bfcc251 : FAILURE in 4m 55s\n- panko-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/0c97c977a48f47a5af6ebf2b73b9aaed : FAILURE in 5m 42s\n- panko-tox-py35-mongodb https://zuul.opendev.org/t/openstack/build/98677077723a4425b19f5559ab9659e7 : TIMED_OUT in 31m 06s\n- panko-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/afb0bd769d314882b5a9862e1cc8f92a : TIMED_OUT in 31m 07s\n- panko-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/6f96d505bdda4e1fb1d659f0d99041b3 : TIMED_OUT in 31m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3e155ffb945f46aab6efd31d1445f455 : FAILURE in 26m 11s","accounts_in_message":[],"_revision_number":1},{"id":"101efea0a0cbeab43fb595e1ab03ed3f351af093","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-05 18:43:17.000000000","message":"Abandoned\n\nThis branch of the project has been moved to End of Life. In order to be able to delete the branch open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf664eb079ea526f4dac167ed2bf7cc574301605","revisions":{"cf664eb079ea526f4dac167ed2bf7cc574301605":{"kind":"REWORK","_number":1,"created":"2021-09-17 09:44:11.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/58/809558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/58/809558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/58/809558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/58/809558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/58/809558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/58/809558/1"}}},"commit":{"parents":[{"commit":"57ef9f1a5daad536ca0e7017ef34cfa1e751de8f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/57ef9f1a5daad536ca0e7017ef34cfa1e751de8f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:37:47.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:44:11.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this project is no longer maintained.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I959ca9795e753051e03be1f9000e50b13517ee72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/cf664eb079ea526f4dac167ed2bf7cc574301605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/cf664eb079ea526f4dac167ed2bf7cc574301605"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
