)]}'
{"id":"openstack%2Fpanko~809559","triplet_id":"openstack%2Fpanko~stable%2Fpike~I959ca9795e753051e03be1f9000e50b13517ee72","project":"openstack/panko","branch":"stable/pike","topic":"remove-c7-tripleo-jobs-stable/stein-stable/rocky-stable/queens-stable/pike","hashtags":[],"change_id":"I959ca9795e753051e03be1f9000e50b13517ee72","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-17 09:44:22.000000000","updated":"2022-08-01 16:12:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809558,"cherry_pick_of_patch_set":1,"meta_rev_id":"70286d26a82b37d490fcad0f66ebff6f272db6ee","_number":809559,"virtual_id_number":809559,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-17 10:30:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 10:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ff99691246cf28831f9d2c3f848734826279541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:44:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/queens.","accounts_in_message":[],"_revision_number":1},{"id":"58e33118f34916563bff37e2268eb56bd763bb84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:30:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc51265b6426472e94c1ab75c73ff0f1 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db554ff6ab9042a7ae7def42ebce8c82 : FAILURE in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4638167028c24b7badbd9a54cfdd0c7a : TIMED_OUT in 41m 12s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/201801c62baf47b4929a9d23a6e81f5b : SUCCESS in 6m 33s\n- panko-tox-py27-mongodb https://zuul.opendev.org/t/openstack/build/9c40a579898b4b29bb4cd410ca028b35 : FAILURE in 5m 37s\n- panko-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/e4e42ec6abf44888bd8350644b9a8c89 : FAILURE in 4m 20s\n- panko-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/00dadeb955c44761b03b528c57bbfb64 : FAILURE in 5m 09s\n- panko-tox-py35-mongodb https://zuul.opendev.org/t/openstack/build/9f73055f3cec4ddbb97227966d646e30 : TIMED_OUT in 31m 21s\n- panko-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/4824c4f2ec9e4ef6a17b30d58fed9081 : TIMED_OUT in 31m 09s\n- panko-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/79d6cae49d264dc4ac7f197151581312 : TIMED_OUT in 30m 58s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/84493ce67e164ff1ab41ead89e06b5a9 : FAILURE in 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"35ba6547098b4003b5711c0ea644f8ef8d8651a3","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-20 06:46:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dbfffb535506c734f924afdccc79a59d182099c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 07:32:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118d5b6d50a747ea9c93bbd244c58d0d : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/736d0f882a8a44d898962fbada8e23f4 : FAILURE in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b3bfdd5c48554b84a8983e00fd50a3da : TIMED_OUT in 41m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7fbaf03170604a2cbb20b49d9a7ddaf6 : SUCCESS in 5m 26s\n- panko-tox-py27-mongodb https://zuul.opendev.org/t/openstack/build/0cf2743c5b8f4257ac20662782e27429 : FAILURE in 5m 16s\n- panko-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/da834c45d28246b9ba8b8b4b3c28a72a : FAILURE in 4m 58s\n- panko-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/45f3445c234e4016b25175cdbd6139f3 : FAILURE in 6m 57s\n- panko-tox-py35-mongodb https://zuul.opendev.org/t/openstack/build/dc6d285ebc81476382baa1e818b10fd6 : TIMED_OUT in 32m 17s\n- panko-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/23650f7f0bd04e7392862c6e94861d0c : TIMED_OUT in 31m 34s\n- panko-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/c36f3eb12d2e478aa22dc5f4eb623cf8 : TIMED_OUT in 31m 33s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/91a7eb084761400cbbf26c757e67a1cd : FAILURE in 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"70286d26a82b37d490fcad0f66ebff6f272db6ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-01 16:12:59.000000000","message":"Abandoned\n\nstable/pike has transitioned to End of Life for panko, open patches need to be abandoned in order to be able to delete the branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3a02e7dbb76e925ed04d1deaece8e556e426c20","revisions":{"c3a02e7dbb76e925ed04d1deaece8e556e426c20":{"kind":"REWORK","_number":1,"created":"2021-09-17 09:44:22.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/59/809559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/59/809559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/59/809559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/59/809559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/59/809559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/59/809559/1"}}},"commit":{"parents":[{"commit":"a5d3b771a84a99a2c271706d82fe0afc50478000","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/a5d3b771a84a99a2c271706d82fe0afc50478000"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:37:47.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:44:22.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this project is no longer maintained.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I959ca9795e753051e03be1f9000e50b13517ee72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/c3a02e7dbb76e925ed04d1deaece8e556e426c20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/c3a02e7dbb76e925ed04d1deaece8e556e426c20"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
