)]}'
{"id":"openstack%2Fpanko~809620","triplet_id":"openstack%2Fpanko~stable%2Fwallaby~I959ca9795e753051e03be1f9000e50b13517ee72","project":"openstack/panko","branch":"stable/wallaby","topic":"remove-c7-tripleo-jobs","hashtags":[],"change_id":"I959ca9795e753051e03be1f9000e50b13517ee72","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-17 09:39:17.000000000","updated":"2024-03-01 09:05:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f6a4199f08be72e8c64557fe04d5eecc2aa90e7","_number":809620,"virtual_id_number":809620,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-17 10:44:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 10:44:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aca79e10a90d4b97789fbd41971c1d85bee889d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 09:39:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"913cc2662d7e908a7ac5f3db083ded897d7372c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:44:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd1efc77ba94b7f91104fd567ecba0f : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b309d61d70504e5d9e85997ded48ff10 : SUCCESS in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/923ad568b7dd46b8a4a54c48db074be9 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a172196f9d594abf84379bfd9022abdf : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2134a23019154116a0d784b24aaec8ed : SUCCESS in 7m 37s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/394ff06b131e4b13ae08dfc0542ea18d : SUCCESS in 48m 13s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/babee0f0cd8344418212aca389ce5523 : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e92171b8002f62ca01ff281d5d74ba0e51646905","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-20 06:49:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa3d2b3f723351d1d423a46e50f1c5cc4291aeba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:51:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/243bac51d1f840829be77d37df2971ca : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a63df1cd66ee463e89b97c685aac632d : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcf81b81ac5647d0a166a9c0c580ea66 : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49fa9001ec464aeab4182a163e8bb496 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd39fbc9d69b4c02a8a3596cba798bbe : SUCCESS in 8m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/18532acd788c422cb57a21da76f4ba62 : SUCCESS in 55m 19s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6a908dbdea554dd5af787d384ef174ee : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5f6a4199f08be72e8c64557fe04d5eecc2aa90e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-03-01 09:05:28.000000000","message":"Abandoned\n\nAbandoned as part of transition from stable to unmaintained. Please repropose to unmaintained branch after creation if still valid interest.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07908b985aa9714965aa7c5fcae64ca7cd898a8c","revisions":{"07908b985aa9714965aa7c5fcae64ca7cd898a8c":{"kind":"REWORK","_number":1,"created":"2021-09-17 09:39:17.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/20/809620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/panko","ref":"refs/changes/20/809620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/panko refs/changes/20/809620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/panko refs/changes/20/809620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/panko refs/changes/20/809620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/panko refs/changes/20/809620/1"}}},"commit":{"parents":[{"commit":"276d891675c692c146448180767f97362871b493","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/276d891675c692c146448180767f97362871b493"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:37:47.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 09:39:14.000000000","tz":330},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this project is no longer maintained.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I959ca9795e753051e03be1f9000e50b13517ee72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/07908b985aa9714965aa7c5fcae64ca7cd898a8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/panko/commit/07908b985aa9714965aa7c5fcae64ca7cd898a8c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
