)]}'
{"id":"openstack%2Fpatrole~648255","triplet_id":"openstack%2Fpatrole~master~I180d1133b4b77ee7aeff596899f9d7ad9b6d14fe","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I180d1133b4b77ee7aeff596899f9d7ad9b6d14fe","subject":"Add Patrole gate job for stable/stein","status":"MERGED","created":"2019-03-27 22:12:50.000000000","updated":"2019-05-16 22:50:47.000000000","submitted":"2019-05-16 22:50:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"648255-1558047047940-16b933f0","meta_rev_id":"7e2b92e032ae2c88226f98b72a6292be02903b76","_number":648255,"virtual_id_number":648255,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2019-05-16 22:50:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2019-04-29 00:10:18.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-04-29 00:10:18.000000000","_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-05-16 20:13:36.000000000","_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-03 02:43:01.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-05-16 20:13:36.000000000","updated_by":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"reviewer":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"state":"REVIEWER"},{"updated":"2019-05-16 22:50:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ec98b01dee876c80102f357527af3ec55eeba69","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-27 22:12:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc8672d7a45f2ecf61db67127577419926fcf553","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 23:43:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/1/check/openstack-tox-cover/e08102e/cover/ : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/1/check/openstack-tox-lower-constraints/35fd3a9/ : SUCCESS in 6m 01s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/1/check/openstack-tox-py36/383b7f0/ : SUCCESS in 6m 17s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/1/check/openstack-tox-py37/b7003aa/ : SUCCESS in 7m 14s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/1/check/openstack-tox-pep8/41b4a75/ : SUCCESS in 6m 46s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/1/check/openstack-tox-py27/b699cbe/ : SUCCESS in 6m 03s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/1/check/openstack-tox-py35/9c763f5/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/55/648255/1/check/openstack-tox-docs/a45b260/html/ : SUCCESS in 8m 02s\n- patrole-admin http://logs.openstack.org/55/648255/1/check/patrole-admin/0415687/ : FAILURE in 58m 23s\n- patrole-member http://logs.openstack.org/55/648255/1/check/patrole-member/5c93509/ : SUCCESS in 1h 10m 19s\n- patrole-reader http://logs.openstack.org/55/648255/1/check/patrole-reader/7b9cf27/ : FAILURE in 1h 03m 50s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/1/check/patrole-member-stein/7f94a7b/ : SUCCESS in 1h 05m 42s\n- patrole-member-rocky http://logs.openstack.org/55/648255/1/check/patrole-member-rocky/f53cb9a/ : SUCCESS in 58m 40s\n- patrole-member-queens http://logs.openstack.org/55/648255/1/check/patrole-member-queens/0f93f0d/ : SUCCESS in 1h 02m 19s\n- patrole-member-pike http://logs.openstack.org/55/648255/1/check/patrole-member-pike/8f8f144/ : SUCCESS in 1h 15m 28s\n- patrole-py35-member http://logs.openstack.org/55/648255/1/check/patrole-py35-member/00ad900/ : SUCCESS in 1h 05m 25s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/1/check/patrole-multinode-admin/49bc63f/ : FAILURE in 56m 48s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/1/check/patrole-multinode-member/5c5e677/ : SUCCESS in 1h 03m 23s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/1/check/patrole-extension-admin/776bc7b/ : SUCCESS in 27m 49s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/1/check/patrole-extension-member/ad20810/ : FAILURE in 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7180fc413386e58dbf60e59a78f541594cc2aea","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-03-28 13:43:29.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b8e168510693a69c0eca3d481e90aa4cc8293f21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 15:03:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/1/check/openstack-tox-cover/7d2e2e3/cover/ : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/1/check/openstack-tox-lower-constraints/2fff09d/ : SUCCESS in 5m 29s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/1/check/openstack-tox-py36/e566fa0/ : SUCCESS in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/1/check/openstack-tox-py37/db46174/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/1/check/openstack-tox-pep8/444b4ef/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/1/check/openstack-tox-py27/0c1cb2c/ : SUCCESS in 5m 24s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/1/check/openstack-tox-py35/37561c9/ : SUCCESS in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/55/648255/1/check/openstack-tox-docs/d03be03/html/ : SUCCESS in 4m 30s\n- patrole-admin http://logs.openstack.org/55/648255/1/check/patrole-admin/e69bbc9/ : SUCCESS in 57m 24s\n- patrole-member http://logs.openstack.org/55/648255/1/check/patrole-member/944f4b5/ : SUCCESS in 1h 01m 41s\n- patrole-reader http://logs.openstack.org/55/648255/1/check/patrole-reader/87de6e8/ : FAILURE in 1h 10m 19s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/1/check/patrole-member-stein/833fadd/ : SUCCESS in 58m 01s\n- patrole-member-rocky http://logs.openstack.org/55/648255/1/check/patrole-member-rocky/05bd01e/ : SUCCESS in 1h 00m 13s\n- patrole-member-queens http://logs.openstack.org/55/648255/1/check/patrole-member-queens/76a17b4/ : FAILURE in 1h 01m 46s\n- patrole-member-pike http://logs.openstack.org/55/648255/1/check/patrole-member-pike/f10c66d/ : SUCCESS in 1h 02m 15s\n- patrole-py35-member http://logs.openstack.org/55/648255/1/check/patrole-py35-member/55b7f5f/ : SUCCESS in 1h 06m 55s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/1/check/patrole-multinode-admin/56e4da0/ : FAILURE in 1h 02m 40s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/1/check/patrole-multinode-member/ea63c62/ : SUCCESS in 1h 03m 13s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/1/check/patrole-extension-admin/719897b/ : SUCCESS in 32m 15s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/1/check/patrole-extension-member/c35c311/ : FAILURE in 37m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f516b9498373802622869691f168f749f3107de9","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-03-28 15:16:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7f71dd0e4ef804a3a22bac368151f8a93066a4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 16:48:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/1/check/openstack-tox-cover/38d9239/cover/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/1/check/openstack-tox-lower-constraints/5aadcc9/ : SUCCESS in 5m 48s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/1/check/openstack-tox-py36/5c73270/ : SUCCESS in 6m 06s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/1/check/openstack-tox-py37/c1195ce/ : SUCCESS in 6m 09s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/1/check/openstack-tox-pep8/c647f38/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/1/check/openstack-tox-py27/bf5381f/ : SUCCESS in 5m 13s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/1/check/openstack-tox-py35/04123ea/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/55/648255/1/check/openstack-tox-docs/2a946f9/html/ : SUCCESS in 4m 54s\n- patrole-admin http://logs.openstack.org/55/648255/1/check/patrole-admin/182cb3c/ : SUCCESS in 46m 34s\n- patrole-member http://logs.openstack.org/55/648255/1/check/patrole-member/4ee7402/ : SUCCESS in 1h 07m 30s\n- patrole-reader http://logs.openstack.org/55/648255/1/check/patrole-reader/229b2a7/ : SUCCESS in 1h 02m 38s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/1/check/patrole-member-stein/2d86218/ : SUCCESS in 1h 09m 12s\n- patrole-member-rocky http://logs.openstack.org/55/648255/1/check/patrole-member-rocky/aa05790/ : SUCCESS in 1h 07m 34s\n- patrole-member-queens http://logs.openstack.org/55/648255/1/check/patrole-member-queens/c85956e/ : SUCCESS in 59m 49s\n- patrole-member-pike http://logs.openstack.org/55/648255/1/check/patrole-member-pike/0c314d2/ : SUCCESS in 1h 05m 30s\n- patrole-py35-member http://logs.openstack.org/55/648255/1/check/patrole-py35-member/0b883c0/ : FAILURE in 54m 27s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/1/check/patrole-multinode-admin/d851f6d/ : FAILURE in 1h 14m 21s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/1/check/patrole-multinode-member/55b9b69/ : FAILURE in 54m 30s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/1/check/patrole-extension-admin/e31f6bd/ : SUCCESS in 43m 29s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/1/check/patrole-extension-member/86a506e/ : FAILURE in 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1ba96cba8bd8ace4692e79b4ea4375fa815c458","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-03-29 12:56:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23eacc753dd23e1b7a72fd869ea35fb0287b60ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 14:22:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/1/check/openstack-tox-cover/fcfa5ef/cover/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/1/check/openstack-tox-lower-constraints/85f671f/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/1/check/openstack-tox-py36/0e21211/ : SUCCESS in 5m 20s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/1/check/openstack-tox-py37/caea3f7/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/1/check/openstack-tox-pep8/830be2b/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/1/check/openstack-tox-py27/04074b9/ : SUCCESS in 4m 23s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/1/check/openstack-tox-py35/98299b5/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/55/648255/1/check/openstack-tox-docs/21f35e3/html/ : SUCCESS in 4m 46s\n- patrole-admin http://logs.openstack.org/55/648255/1/check/patrole-admin/e46fa80/ : SUCCESS in 59m 26s\n- patrole-member http://logs.openstack.org/55/648255/1/check/patrole-member/0d7c627/ : SUCCESS in 1h 03m 47s\n- patrole-reader http://logs.openstack.org/55/648255/1/check/patrole-reader/537f6ed/ : SUCCESS in 1h 10m 05s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/1/check/patrole-member-stein/b8e7177/ : SUCCESS in 1h 04m 57s\n- patrole-member-rocky http://logs.openstack.org/55/648255/1/check/patrole-member-rocky/5fa78f7/ : SUCCESS in 1h 07m 06s\n- patrole-member-queens http://logs.openstack.org/55/648255/1/check/patrole-member-queens/f855363/ : SUCCESS in 1h 03m 04s\n- patrole-member-pike http://logs.openstack.org/55/648255/1/check/patrole-member-pike/558e427/ : SUCCESS in 1h 01m 33s\n- patrole-py35-member http://logs.openstack.org/55/648255/1/check/patrole-py35-member/8711ac2/ : SUCCESS in 1h 12m 52s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/1/check/patrole-multinode-admin/a25a57d/ : FAILURE in 1h 00m 42s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/1/check/patrole-multinode-member/f6c27a7/ : SUCCESS in 56m 55s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/1/check/patrole-extension-admin/5d0d38e/ : SUCCESS in 31m 15s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/1/check/patrole-extension-member/ea943f5/ : FAILURE in 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6def963daa9025ee34d53ad2bb1df5fdd62ba172","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-04-03 02:43:01.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks for taking the initiative. I was going to get to this if not you!","accounts_in_message":[],"_revision_number":1},{"id":"958ded8a67095b9d312d600d3d29f3b27784ca6c","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-04-03 02:43:24.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"092ad57fdfd10dd30b580050fe8c52a249928095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 02:43:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05a38a3e6b517e3d7de8fcd2189084af2e2f8f3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 03:51:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/1/gate/openstack-tox-lower-constraints/48846ae/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/1/gate/openstack-tox-py36/94a9309/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/1/gate/openstack-tox-py37/43e9a81/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/1/gate/openstack-tox-pep8/4565081/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/1/gate/openstack-tox-py27/19353fb/ : SUCCESS in 4m 00s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/1/gate/openstack-tox-py35/fa879ad/ : SUCCESS in 4m 42s\n- openstack-tox-docs http://logs.openstack.org/55/648255/1/gate/openstack-tox-docs/b076668/html/ : SUCCESS in 4m 00s\n- patrole-admin http://logs.openstack.org/55/648255/1/gate/patrole-admin/357616b/ : FAILURE in 55m 03s\n- patrole-member http://logs.openstack.org/55/648255/1/gate/patrole-member/85b1505/ : FAILURE in 1h 02m 18s","accounts_in_message":[],"_revision_number":1},{"id":"30527266443757e6305f753fd281fbc836f4c108","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-04-03 14:03:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b6814071b4b1b016e1353fd0805a3f86f4db596","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 15:48:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/1/check/openstack-tox-cover/c8e9d56/cover/ : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/1/check/openstack-tox-lower-constraints/4bc5821/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/1/check/openstack-tox-py36/7279ec7/ : SUCCESS in 5m 30s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/1/check/openstack-tox-py37/3ba174f/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/1/check/openstack-tox-pep8/692bedc/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/1/check/openstack-tox-py27/cd42768/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/1/check/openstack-tox-py35/61f3235/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/55/648255/1/check/openstack-tox-docs/5576a2e/html/ : SUCCESS in 5m 57s\n- patrole-admin http://logs.openstack.org/55/648255/1/check/patrole-admin/e81bb83/ : FAILURE in 1h 03m 54s\n- patrole-member http://logs.openstack.org/55/648255/1/check/patrole-member/94a6ea8/ : FAILURE in 1h 13m 07s\n- patrole-reader http://logs.openstack.org/55/648255/1/check/patrole-reader/88b8b4d/ : FAILURE in 1h 03m 38s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/1/check/patrole-member-stein/482c1bf/ : SUCCESS in 1h 38m 21s\n- patrole-member-rocky http://logs.openstack.org/55/648255/1/check/patrole-member-rocky/7f9b3d9/ : SUCCESS in 1h 06m 37s\n- patrole-member-queens http://logs.openstack.org/55/648255/1/check/patrole-member-queens/a2c3a9b/ : SUCCESS in 1h 02m 07s\n- patrole-member-pike http://logs.openstack.org/55/648255/1/check/patrole-member-pike/97bdb2c/ : SUCCESS in 57m 07s\n- patrole-py35-member http://logs.openstack.org/55/648255/1/check/patrole-py35-member/231971f/ : FAILURE in 1h 08m 32s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/1/check/patrole-multinode-admin/5b4f5a6/ : FAILURE in 1h 08m 56s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/1/check/patrole-multinode-member/9c1054f/ : FAILURE in 1h 04m 22s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/1/check/patrole-extension-admin/e6a1355/ : FAILURE in 35m 35s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/1/check/patrole-extension-member/57bff57/ : FAILURE in 36m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7defb794e49863c04afe783b1ef07990907d2a86","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-04-03 17:12:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c8e33f7e7841d6a85d21ffe46d3740381e6cf73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 18:36:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/1/check/openstack-tox-cover/88c1f86/cover/ : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/1/check/openstack-tox-lower-constraints/39436a2/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/1/check/openstack-tox-py36/035c787/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/1/check/openstack-tox-py37/da33b3e/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/1/check/openstack-tox-pep8/219403c/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/1/check/openstack-tox-py27/9d92eb8/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/1/check/openstack-tox-py35/0279afa/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/55/648255/1/check/openstack-tox-docs/2834fd5/html/ : SUCCESS in 4m 49s\n- patrole-admin http://logs.openstack.org/55/648255/1/check/patrole-admin/7dbc382/ : FAILURE in 55m 44s\n- patrole-member http://logs.openstack.org/55/648255/1/check/patrole-member/55f1f9e/ : FAILURE in 1h 11m 16s\n- patrole-reader http://logs.openstack.org/55/648255/1/check/patrole-reader/d42cdb8/ : FAILURE in 1h 15m 46s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/1/check/patrole-member-stein/fed6802/ : SUCCESS in 1h 14m 10s\n- patrole-member-rocky http://logs.openstack.org/55/648255/1/check/patrole-member-rocky/5f09684/ : FAILURE in 1h 08m 06s\n- patrole-member-queens http://logs.openstack.org/55/648255/1/check/patrole-member-queens/e6764bc/ : SUCCESS in 1h 01m 39s\n- patrole-member-pike http://logs.openstack.org/55/648255/1/check/patrole-member-pike/96dbc71/ : FAILURE in 1h 05m 04s\n- patrole-py35-member http://logs.openstack.org/55/648255/1/check/patrole-py35-member/12c51d7/ : FAILURE in 1h 05m 16s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/1/check/patrole-multinode-admin/c281c21/ : FAILURE in 1h 06m 37s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/1/check/patrole-multinode-member/7219d76/ : FAILURE in 55m 06s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/1/check/patrole-extension-admin/074256e/ : FAILURE in 37m 31s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/1/check/patrole-extension-member/74505fc/ : FAILURE in 32m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6d65bc0730c2d56277faf6c82a14513e899721f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-03 19:01:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"09a52214e9e913104ab526b7830ec7adaca52105","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-04-03 19:21:55.000000000","message":"Patch Set 1:\n\nIt fails several times with same reason, so I think there is an issue with the code","accounts_in_message":[],"_revision_number":1},{"id":"6959d2f1372bb6ccecec6be808966cb87b2475d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 20:26:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/1/check/openstack-tox-cover/192ffc1/cover/ : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/1/check/openstack-tox-lower-constraints/fce36cf/ : SUCCESS in 5m 55s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/1/check/openstack-tox-py36/f02d300/ : SUCCESS in 6m 40s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/1/check/openstack-tox-py37/ac41ac8/ : SUCCESS in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/1/check/openstack-tox-pep8/a6f4cbc/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/1/check/openstack-tox-py27/4e12b28/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/1/check/openstack-tox-py35/14539f9/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/55/648255/1/check/openstack-tox-docs/e1961fa/html/ : SUCCESS in 5m 43s\n- patrole-admin http://logs.openstack.org/55/648255/1/check/patrole-admin/63b6639/ : FAILURE in 47m 24s\n- patrole-member http://logs.openstack.org/55/648255/1/check/patrole-member/afff421/ : FAILURE in 1h 03m 53s\n- patrole-reader http://logs.openstack.org/55/648255/1/check/patrole-reader/1434265/ : FAILURE in 1h 01m 59s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/1/check/patrole-member-stein/bf06a50/ : FAILURE in 1h 09m 38s\n- patrole-member-rocky http://logs.openstack.org/55/648255/1/check/patrole-member-rocky/e5db919/ : SUCCESS in 1h 15m 26s\n- patrole-member-queens http://logs.openstack.org/55/648255/1/check/patrole-member-queens/cfdad8b/ : SUCCESS in 1h 00m 15s\n- patrole-member-pike http://logs.openstack.org/55/648255/1/check/patrole-member-pike/9d8a382/ : FAILURE in 1h 02m 05s\n- patrole-py35-member http://logs.openstack.org/55/648255/1/check/patrole-py35-member/f9eb88b/ : FAILURE in 1h 09m 42s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/1/check/patrole-multinode-admin/640199f/ : FAILURE in 1h 16m 54s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/1/check/patrole-multinode-member/25a3b79/ : FAILURE in 1h 05m 30s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/1/check/patrole-extension-admin/c6f561a/ : FAILURE in 30m 14s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/1/check/patrole-extension-member/18068d9/ : FAILURE in 43m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b062f13d8921ae8907ec27e048d86d0cfc170b82","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-04 14:55:35.000000000","message":"Patch Set 1:\n\n\u003e It fails several times with same reason, so I think there is an\n \u003e issue with the code\n\nyeah, its shared network creation issue.","accounts_in_message":[],"_revision_number":1},{"id":"7a74e58f7787e1ea703cece11717e34cca0912ef","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-04-29 00:10:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"eae6f566ddedd2f9491e226d4c93daa418d9d6c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 01:36:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/2/check/openstack-tox-cover/f48396d/cover/ : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/2/check/openstack-tox-lower-constraints/79bcb9e/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/2/check/openstack-tox-py36/2c63433/ : SUCCESS in 3m 58s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/2/check/openstack-tox-py37/8f85ec9/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/2/check/openstack-tox-pep8/6f6d02c/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/2/check/openstack-tox-py27/4a3ccea/ : SUCCESS in 4m 23s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/2/check/openstack-tox-py35/5f6b402/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/55/648255/2/check/openstack-tox-docs/7a5080a/html/ : SUCCESS in 3m 44s\n- patrole-admin http://logs.openstack.org/55/648255/2/check/patrole-admin/44f6551/ : SUCCESS in 57m 57s\n- patrole-member http://logs.openstack.org/55/648255/2/check/patrole-member/4f01ad4/ : FAILURE in 1h 10m 37s\n- patrole-reader http://logs.openstack.org/55/648255/2/check/patrole-reader/126dd1b/ : SUCCESS in 1h 05m 02s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/2/check/patrole-member-stein/925a535/ : FAILURE in 56m 27s\n- patrole-member-rocky http://logs.openstack.org/55/648255/2/check/patrole-member-rocky/c30cdd4/ : SUCCESS in 1h 07m 28s\n- patrole-member-queens http://logs.openstack.org/55/648255/2/check/patrole-member-queens/211dbdf/ : SUCCESS in 1h 01m 56s\n- patrole-member-pike http://logs.openstack.org/55/648255/2/check/patrole-member-pike/c0db340/ : SUCCESS in 1h 10m 24s\n- patrole-py35-member http://logs.openstack.org/55/648255/2/check/patrole-py35-member/85d05e7/ : SUCCESS in 1h 02m 05s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/2/check/patrole-multinode-admin/a00d03e/ : FAILURE in 1h 17m 54s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/2/check/patrole-multinode-member/94f5508/ : FAILURE in 30m 02s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/2/check/patrole-extension-admin/f40d334/ : FAILURE in 29m 23s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/2/check/patrole-extension-member/7bea17f/ : FAILURE in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6cd55a542bf5c202c10812671844a6c8d97155b","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-04-29 14:02:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"baf32a834a5d21022d4c4f8a50d33dceaf92ab1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 15:44:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/2/check/openstack-tox-cover/a22975d/cover/ : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/2/check/openstack-tox-lower-constraints/779bb8c/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/2/check/openstack-tox-py36/7824edc/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/2/check/openstack-tox-py37/6212912/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/2/check/openstack-tox-pep8/6f88d5c/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/2/check/openstack-tox-py27/bf2f3fe/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/55/648255/2/check/openstack-tox-py35/6a70690/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/55/648255/2/check/openstack-tox-docs/d751e4b/html/ : SUCCESS in 5m 45s\n- patrole-admin http://logs.openstack.org/55/648255/2/check/patrole-admin/8815c2c/ : SUCCESS in 56m 18s\n- patrole-member http://logs.openstack.org/55/648255/2/check/patrole-member/90c9236/ : SUCCESS in 1h 20m 43s\n- patrole-reader http://logs.openstack.org/55/648255/2/check/patrole-reader/16abbb7/ : SUCCESS in 1h 13m 20s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/2/check/patrole-member-stein/f3d8f3f/ : FAILURE in 59m 31s\n- patrole-member-rocky http://logs.openstack.org/55/648255/2/check/patrole-member-rocky/0595c82/ : SUCCESS in 1h 13m 25s\n- patrole-member-queens http://logs.openstack.org/55/648255/2/check/patrole-member-queens/aa92553/ : SUCCESS in 1h 16m 54s\n- patrole-member-pike http://logs.openstack.org/55/648255/2/check/patrole-member-pike/2280234/ : SUCCESS in 1h 08m 18s\n- patrole-py35-member http://logs.openstack.org/55/648255/2/check/patrole-py35-member/23c412f/ : SUCCESS in 1h 06m 39s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/2/check/patrole-multinode-admin/8d4fed8/ : FAILURE in 1h 09m 07s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/2/check/patrole-multinode-member/6a55e5b/ : SUCCESS in 1h 07m 39s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/2/check/patrole-extension-admin/e2faba3/ : FAILURE in 29m 18s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/2/check/patrole-extension-member/8a4822d/ : POST_FAILURE in 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"de18a9feec7b02bb938fa3ec5392e81e46b9ffa6","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-05-03 22:35:15.000000000","message":"Patch Set 2:\n\nNeed to skip a few more of these networking tests, I\u0027ll push a PS later to do that.","accounts_in_message":[],"_revision_number":2},{"id":"f11ff9c91aecce2b826713041de058f9d15cd84b","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-05-10 14:32:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6ec4e5df99544ffb70575b56fd081134160b7c70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:02:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/2/check/openstack-tox-cover/7262316/cover/ : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/2/check/openstack-tox-lower-constraints/03733fa/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/2/check/openstack-tox-py36/01e29cf/ : SUCCESS in 4m 30s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/2/check/openstack-tox-py37/86c9beb/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/2/check/openstack-tox-pep8/59beda4/ : FAILURE in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/2/check/openstack-tox-py27/a701231/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/55/648255/2/check/openstack-tox-docs/9571f9c/html/ : SUCCESS in 3m 34s\n- patrole-admin http://logs.openstack.org/55/648255/2/check/patrole-admin/9511ee2/ : FAILURE in 50m 23s\n- patrole-member http://logs.openstack.org/55/648255/2/check/patrole-member/cd48973/ : SUCCESS in 1h 21m 27s\n- patrole-reader http://logs.openstack.org/55/648255/2/check/patrole-reader/8d0e9ba/ : SUCCESS in 1h 19m 47s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/2/check/patrole-member-stein/28a20d4/ : SUCCESS in 1h 09m 02s\n- patrole-member-rocky http://logs.openstack.org/55/648255/2/check/patrole-member-rocky/ca46b4d/ : SUCCESS in 1h 08m 26s\n- patrole-member-queens http://logs.openstack.org/55/648255/2/check/patrole-member-queens/417b96a/ : SUCCESS in 1h 08m 04s\n- patrole-member-pike http://logs.openstack.org/55/648255/2/check/patrole-member-pike/882038d/ : SUCCESS in 1h 09m 10s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/2/check/patrole-multinode-admin/14ca109/ : FAILURE in 1h 12m 11s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/2/check/patrole-multinode-member/e954a21/ : FAILURE in 1h 24m 39s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/2/check/patrole-extension-admin/e5f67d4/ : FAILURE in 31m 54s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/2/check/patrole-extension-member/ebf416f/ : FAILURE in 37m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57d2720a67360aee564b777a20987052f38a8692","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-05-16 20:13:36.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"84df04fd41de9612772fb1638c06a642a1a698d9","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-05-16 20:13:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7f28f141c6e28778a5620c4a8b74182f2e80d664","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 21:38:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/55/648255/2/check/openstack-tox-cover/ddec896/cover/ : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/2/check/openstack-tox-lower-constraints/3b857fb/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/2/check/openstack-tox-py36/064ddd7/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/2/check/openstack-tox-py37/0597fda/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/2/check/openstack-tox-pep8/7a32db0/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/2/check/openstack-tox-py27/bd20b23/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/55/648255/2/check/openstack-tox-docs/ffa582d/html/ : SUCCESS in 4m 08s\n- patrole-admin http://logs.openstack.org/55/648255/2/check/patrole-admin/9e8113b/ : SUCCESS in 55m 28s\n- patrole-member http://logs.openstack.org/55/648255/2/check/patrole-member/fbc02cd/ : SUCCESS in 1h 07m 12s\n- patrole-reader http://logs.openstack.org/55/648255/2/check/patrole-reader/319b2fb/ : SUCCESS in 1h 07m 30s (non-voting)\n- patrole-member-stein http://logs.openstack.org/55/648255/2/check/patrole-member-stein/64d3982/ : SUCCESS in 1h 04m 25s\n- patrole-member-rocky http://logs.openstack.org/55/648255/2/check/patrole-member-rocky/f3c0bc5/ : SUCCESS in 1h 14m 28s\n- patrole-member-queens http://logs.openstack.org/55/648255/2/check/patrole-member-queens/b2d5010/ : SUCCESS in 58m 37s\n- patrole-member-pike http://logs.openstack.org/55/648255/2/check/patrole-member-pike/3fb79d3/ : SUCCESS in 1h 10m 43s\n- patrole-multinode-admin http://logs.openstack.org/55/648255/2/check/patrole-multinode-admin/5f7fd27/ : FAILURE in 1h 11m 30s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/55/648255/2/check/patrole-multinode-member/d8fe3ad/ : SUCCESS in 1h 04m 02s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/55/648255/2/check/patrole-extension-admin/d914d18/ : FAILURE in 28m 53s (non-voting)\n- patrole-extension-member http://logs.openstack.org/55/648255/2/check/patrole-extension-member/fd1eef7/ : FAILURE in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"523a5b079ba1c26f6ec880f3ffebf528313760a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 21:38:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f0c84f6de637ce18c5ec0d72b414719ddbeb1d24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:50:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7e2b92e032ae2c88226f98b72a6292be02903b76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:50:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/55/648255/2/gate/openstack-tox-lower-constraints/be42d66/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/55/648255/2/gate/openstack-tox-py36/e58145c/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/55/648255/2/gate/openstack-tox-py37/c24f6e5/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/55/648255/2/gate/openstack-tox-pep8/dca29f4/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/55/648255/2/gate/openstack-tox-py27/fee878c/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/55/648255/2/gate/openstack-tox-docs/2ab184d/html/ : SUCCESS in 3m 37s\n- patrole-admin http://logs.openstack.org/55/648255/2/gate/patrole-admin/7fbb078/ : SUCCESS in 1h 06m 03s\n- patrole-member http://logs.openstack.org/55/648255/2/gate/patrole-member/fd28f8a/ : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b3606a058bb03a857998d097a2a1b29b4daf27bb","revisions":{"ee2f6c82a20e8f1f7ff62706a70fa19beee961f2":{"kind":"REWORK","_number":1,"created":"2019-03-27 22:12:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/55/648255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/55/648255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/648255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/648255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/648255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/55/648255/1"}}},"commit":{"parents":[{"commit":"f04d671d15f2f04a0ab9172ce89e668ba455ac59","subject":"Add release notes page for v0.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f04d671d15f2f04a0ab9172ce89e668ba455ac59"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-03-27 22:11:45.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-03-27 22:11:45.000000000","tz":0},"subject":"Add Patrole gate job for stable/stein","message":"Add Patrole gate job for stable/stein\n\nNow we have stable/stein branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Project testing for stable/stein\nby adding new jobs running on stable/stein version of\nopenstack.\nDepends-On: https://review.openstack.org/#/c/647867/\n\nChange-Id: I180d1133b4b77ee7aeff596899f9d7ad9b6d14fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ee2f6c82a20e8f1f7ff62706a70fa19beee961f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ee2f6c82a20e8f1f7ff62706a70fa19beee961f2"}]},"branch":"refs/heads/master"},"b3606a058bb03a857998d097a2a1b29b4daf27bb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-29 00:10:18.000000000","uploader":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"ref":"refs/changes/55/648255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/55/648255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/648255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/648255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/55/648255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/55/648255/2"}}},"commit":{"parents":[{"commit":"4e081dfd68b3e646c67c270438c340482a98e907","subject":"Merge \"make Patrole to run for shared network env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4e081dfd68b3e646c67c270438c340482a98e907"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-03-27 22:11:45.000000000","tz":0},"committer":{"name":"Felipe Monteiro","email":"felipe.monteiro@att.com","date":"2019-04-29 00:10:18.000000000","tz":0},"subject":"Add Patrole gate job for stable/stein","message":"Add Patrole gate job for stable/stein\n\nNow we have stable/stein branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Project testing for stable/stein\nby adding new jobs running on stable/stein version of\nopenstack.\nDepends-On: https://review.openstack.org/#/c/647867/\n\nChange-Id: I180d1133b4b77ee7aeff596899f9d7ad9b6d14fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/b3606a058bb03a857998d097a2a1b29b4daf27bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/b3606a058bb03a857998d097a2a1b29b4daf27bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
