)]}'
{"id":"openstack%2Fpatrole~654293","triplet_id":"openstack%2Fpatrole~master~I09222df09e6c6618a06a08479638be3953e19ed4","project":"openstack/patrole","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"I09222df09e6c6618a06a08479638be3953e19ed4","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-22 03:57:52.000000000","updated":"2019-04-29 15:19:04.000000000","submitted":"2019-04-29 15:19:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654293-1556551144895-5f13658a","meta_rev_id":"c616603250a1e21c579badedcbe401574e1903bd","_number":654293,"virtual_id_number":654293,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2019-04-29 15:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2019-04-29 00:09:55.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-29 00:09:55.000000000","_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":1,"date":"2019-04-29 14:02:17.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 15:18:47.000000000","updated_by":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"reviewer":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"state":"REVIEWER"},{"updated":"2019-04-29 14:02:17.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-04-29 15:19:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6df61a9d600a146f549f53cfcad507060afc7f20","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-22 03:57:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae5915c2dac63c4ee77eef992baebe214e7a351f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 05:12:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/93/654293/1/check/test-release-openstack/158ff8a/ : SUCCESS in 4m 06s\n- openstack-tox-cover http://logs.openstack.org/93/654293/1/check/openstack-tox-cover/7544764/cover/ : SUCCESS in 7m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/654293/1/check/openstack-tox-lower-constraints/119d323/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/93/654293/1/check/openstack-tox-py36/5d4c3bd/ : SUCCESS in 5m 05s\n- openstack-tox-py37 http://logs.openstack.org/93/654293/1/check/openstack-tox-py37/271264f/ : SUCCESS in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/93/654293/1/check/openstack-tox-pep8/d1e54f5/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/93/654293/1/check/openstack-tox-py27/e1677dd/ : SUCCESS in 4m 22s\n- openstack-tox-docs http://logs.openstack.org/93/654293/1/check/openstack-tox-docs/6d9992e/html/ : SUCCESS in 3m 57s\n- build-openstack-releasenotes http://logs.openstack.org/93/654293/1/check/build-openstack-releasenotes/dc2db9e/html/ : SUCCESS in 4m 44s\n- patrole-admin http://logs.openstack.org/93/654293/1/check/patrole-admin/8555d4a/ : FAILURE in 53m 22s\n- patrole-member http://logs.openstack.org/93/654293/1/check/patrole-member/f1059fe/ : FAILURE in 1h 08m 56s\n- patrole-reader http://logs.openstack.org/93/654293/1/check/patrole-reader/09b8d32/ : FAILURE in 1h 04m 34s (non-voting)\n- patrole-member-rocky http://logs.openstack.org/93/654293/1/check/patrole-member-rocky/99c0627/ : SUCCESS in 1h 11m 44s\n- patrole-member-queens http://logs.openstack.org/93/654293/1/check/patrole-member-queens/0c8f65f/ : SUCCESS in 1h 02m 15s\n- patrole-member-pike http://logs.openstack.org/93/654293/1/check/patrole-member-pike/c9be95c/ : SUCCESS in 1h 08m 38s\n- patrole-multinode-admin http://logs.openstack.org/93/654293/1/check/patrole-multinode-admin/c3052d3/ : FAILURE in 1h 07m 58s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/93/654293/1/check/patrole-multinode-member/b88cbfd/ : FAILURE in 54m 52s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/93/654293/1/check/patrole-extension-admin/671f083/ : FAILURE in 30m 56s (non-voting)\n- patrole-extension-member http://logs.openstack.org/93/654293/1/check/patrole-extension-member/7573db2/ : FAILURE in 30m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ee95f559e7974fe81dfbb558a5859fe74d5394b","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-04-22 15:18:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThe changes are trivial, so +1 for workflow","accounts_in_message":[],"_revision_number":1},{"id":"534ceb8ba9333bf3dfce76dabef4091ac4c7fa9d","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-04-22 15:18:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"883d3d05e774a17f21eea8111d8796e7dd85b803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:36:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/93/654293/1/check/test-release-openstack/edc353b/ : SUCCESS in 3m 20s\n- openstack-tox-cover http://logs.openstack.org/93/654293/1/check/openstack-tox-cover/e37350a/cover/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/654293/1/check/openstack-tox-lower-constraints/dad6fef/ : SUCCESS in 6m 26s\n- openstack-tox-py36 http://logs.openstack.org/93/654293/1/check/openstack-tox-py36/6da0bd3/ : SUCCESS in 6m 23s\n- openstack-tox-py37 http://logs.openstack.org/93/654293/1/check/openstack-tox-py37/0c56426/ : SUCCESS in 6m 50s\n- openstack-tox-pep8 http://logs.openstack.org/93/654293/1/check/openstack-tox-pep8/6de3a23/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/93/654293/1/check/openstack-tox-py27/21ddab8/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/93/654293/1/check/openstack-tox-docs/2ec7605/html/ : SUCCESS in 4m 32s\n- build-openstack-releasenotes http://logs.openstack.org/93/654293/1/check/build-openstack-releasenotes/38ed484/html/ : SUCCESS in 3m 52s\n- patrole-admin http://logs.openstack.org/93/654293/1/check/patrole-admin/03e3d66/ : FAILURE in 58m 55s\n- patrole-member http://logs.openstack.org/93/654293/1/check/patrole-member/b873c25/ : FAILURE in 1h 04m 41s\n- patrole-reader http://logs.openstack.org/93/654293/1/check/patrole-reader/46691ff/ : FAILURE in 1h 06m 17s (non-voting)\n- patrole-member-rocky http://logs.openstack.org/93/654293/1/check/patrole-member-rocky/acf5391/ : SUCCESS in 1h 01m 59s\n- patrole-member-queens http://logs.openstack.org/93/654293/1/check/patrole-member-queens/50abc68/ : SUCCESS in 50m 41s\n- patrole-member-pike http://logs.openstack.org/93/654293/1/check/patrole-member-pike/874a64b/ : SUCCESS in 1h 05m 07s\n- patrole-multinode-admin http://logs.openstack.org/93/654293/1/check/patrole-multinode-admin/37e4177/ : FAILURE in 1h 07m 02s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/93/654293/1/check/patrole-multinode-member/c1272eb/ : FAILURE in 1h 03m 34s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/93/654293/1/check/patrole-extension-admin/1f475a1/ : FAILURE in 32m 56s (non-voting)\n- patrole-extension-member http://logs.openstack.org/93/654293/1/check/patrole-extension-member/6746006/ : FAILURE in 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04e645b7b108bcb90e711c8072951542a2799a3d","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-04-23 00:45:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"58389a5679981cc1484bc149bdbd41a0f6f31930","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-04-29 00:09:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"63d4876296b2315738442144eec42eace3f74606","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 01:53:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/93/654293/2/check/test-release-openstack/1bd8a40/ : SUCCESS in 2m 59s\n- openstack-tox-cover http://logs.openstack.org/93/654293/2/check/openstack-tox-cover/84de16d/cover/ : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/654293/2/check/openstack-tox-lower-constraints/422e749/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/93/654293/2/check/openstack-tox-py36/bb8d201/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/93/654293/2/check/openstack-tox-py37/4991cba/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/93/654293/2/check/openstack-tox-pep8/c4530db/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/93/654293/2/check/openstack-tox-py27/4c0d932/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/93/654293/2/check/openstack-tox-docs/1dc959f/html/ : SUCCESS in 4m 53s\n- build-openstack-releasenotes http://logs.openstack.org/93/654293/2/check/build-openstack-releasenotes/455b908/html/ : SUCCESS in 3m 30s\n- patrole-admin http://logs.openstack.org/93/654293/2/check/patrole-admin/bab7f1f/ : SUCCESS in 55m 41s\n- patrole-member http://logs.openstack.org/93/654293/2/check/patrole-member/954e00a/ : SUCCESS in 1h 12m 57s\n- patrole-reader http://logs.openstack.org/93/654293/2/check/patrole-reader/82d601d/ : SUCCESS in 1h 08m 06s (non-voting)\n- patrole-member-rocky http://logs.openstack.org/93/654293/2/check/patrole-member-rocky/183f744/ : SUCCESS in 1h 19m 49s\n- patrole-member-queens http://logs.openstack.org/93/654293/2/check/patrole-member-queens/f8d127c/ : SUCCESS in 1h 07m 23s\n- patrole-member-pike http://logs.openstack.org/93/654293/2/check/patrole-member-pike/c882b28/ : SUCCESS in 1h 02m 42s\n- patrole-multinode-admin http://logs.openstack.org/93/654293/2/check/patrole-multinode-admin/1fdc6de/ : FAILURE in 1h 10m 26s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/93/654293/2/check/patrole-multinode-member/17a9114/ : SUCCESS in 1h 05m 46s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/93/654293/2/check/patrole-extension-admin/b98c225/ : FAILURE in 26m 54s (non-voting)\n- patrole-extension-member http://logs.openstack.org/93/654293/2/check/patrole-extension-member/39e7e62/ : FAILURE in 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00e1efbcab501f46b462194d0600e1c8016554bb","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-04-29 14:02:17.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"38c30efa04d70fa0dda1627fbbbd8839d39024fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:04:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6759e3be896e12319d9329620f4e40d850677e7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 15:19:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/93/654293/2/gate/test-release-openstack/fb79ed0/ : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/654293/2/gate/openstack-tox-lower-constraints/4b51638/ : SUCCESS in 8m 27s\n- openstack-tox-py36 http://logs.openstack.org/93/654293/2/gate/openstack-tox-py36/be70c39/ : SUCCESS in 3m 54s\n- openstack-tox-py37 http://logs.openstack.org/93/654293/2/gate/openstack-tox-py37/c235613/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/93/654293/2/gate/openstack-tox-pep8/bc9bf60/ : SUCCESS in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/93/654293/2/gate/openstack-tox-py27/68c3902/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/93/654293/2/gate/openstack-tox-docs/fe1425b/html/ : SUCCESS in 3m 59s\n- build-openstack-releasenotes http://logs.openstack.org/93/654293/2/gate/build-openstack-releasenotes/2e8b6ab/html/ : SUCCESS in 3m 36s\n- patrole-admin http://logs.openstack.org/93/654293/2/gate/patrole-admin/b9184e0/ : SUCCESS in 59m 50s\n- patrole-member http://logs.openstack.org/93/654293/2/gate/patrole-member/6f4ac93/ : SUCCESS in 1h 08m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a8a818bb67083a01fa6b6bc6c57291b67b780c74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 15:19:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"35062ad4152901fb04bb67a9a65dd0b3247ec308","revisions":{"a2aff763488bdc8ada0ccd5d0702c0b99d9851be":{"kind":"REWORK","_number":1,"created":"2019-04-22 03:57:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/654293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/93/654293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/93/654293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/93/654293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/93/654293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/93/654293/1"}}},"commit":{"parents":[{"commit":"eedb039d5b09649228fb79deffb58310fd02fb9e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/eedb039d5b09649228fb79deffb58310fd02fb9e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-22 03:57:35.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-22 03:57:35.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I09222df09e6c6618a06a08479638be3953e19ed4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/a2aff763488bdc8ada0ccd5d0702c0b99d9851be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/a2aff763488bdc8ada0ccd5d0702c0b99d9851be"}]},"branch":"refs/heads/master"},"35062ad4152901fb04bb67a9a65dd0b3247ec308":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-29 00:09:55.000000000","uploader":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"ref":"refs/changes/93/654293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/93/654293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/93/654293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/93/654293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/93/654293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/93/654293/2"}}},"commit":{"parents":[{"commit":"4e081dfd68b3e646c67c270438c340482a98e907","subject":"Merge \"make Patrole to run for shared network env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4e081dfd68b3e646c67c270438c340482a98e907"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-22 03:57:35.000000000","tz":0},"committer":{"name":"Felipe Monteiro","email":"felipe.monteiro@att.com","date":"2019-04-29 00:09:55.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I09222df09e6c6618a06a08479638be3953e19ed4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/35062ad4152901fb04bb67a9a65dd0b3247ec308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/35062ad4152901fb04bb67a9a65dd0b3247ec308"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
