)]}'
{"id":"openstack%2Fpatrole~658775","triplet_id":"openstack%2Fpatrole~master~Ie43108de22ea749990750f7f20d4d66475d33547","project":"openstack/patrole","branch":"master","topic":"broken_url","hashtags":[],"change_id":"Ie43108de22ea749990750f7f20d4d66475d33547","subject":"Fix some inaccurate URL links","status":"MERGED","created":"2019-05-13 09:37:13.000000000","updated":"2019-05-16 22:50:48.000000000","submitted":"2019-05-16 22:50:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658775-1558047048649-a5bd43bf","meta_rev_id":"3c4825be141ce4b8dfa1eefe98231014a9c6e9d3","_number":658775,"virtual_id_number":658775,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2019-05-16 22:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-15 01:42:37.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2019-05-16 22:16:50.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":2,"date":"2019-05-16 20:42:19.000000000","_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":1,"date":"2019-05-16 22:16:50.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 20:42:19.000000000","updated_by":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"reviewer":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"state":"REVIEWER"},{"updated":"2019-05-16 22:16:50.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-05-16 22:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a4d7156eb801cb2bbf7abc7091ad8bb9fe84d19","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-05-13 09:37:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"327977bcb1477092ca611fd8516f90c68e4a4efe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 09:42:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/75/658775/1/check/openstack-tox-pep8/0ae06a5/ : FAILURE in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/75/658775/1/check/openstack-tox-docs/098c30d/html/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"44fc9daeb41f0ca45a74a2d32320997bfb179af2","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-05-14 00:58:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76fda203bb47457a462cc8d3905e175e4d6fe907","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:03:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/75/658775/1/check/openstack-tox-pep8/66fcaf1/ : FAILURE in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/75/658775/1/check/openstack-tox-docs/eb1b128/html/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"468e92c3540eba7588321feb01ffb50af930a7f5","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-05-15 01:18:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd5a798d6cce008f106c5a93a54d8c2748ff31c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 01:36:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/75/658775/1/check/openstack-tox-pep8/84a74fd/ : FAILURE in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/75/658775/1/check/openstack-tox-docs/0ca2c5b/html/ : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0767317f7c9e2c90b4c8f12200b1adddfb074deb","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-05-15 01:42:37.000000000","message":"Patch Set 1:\n\nI don\u0027t know why pep8 fails...","accounts_in_message":[],"_revision_number":1},{"id":"380d4c66901a5da86a852e896b32b5e1da9a7d88","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-05-16 20:13:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee05d915a7adcda22e4837054842e391bf25159e","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-05-16 20:14:44.000000000","message":"Patch Set 1:\n\n@zhufl, I fixed pep8 job: https://review.opendev.org/#/c/658811/","accounts_in_message":[],"_revision_number":1},{"id":"f61b1ef83217f99ff4f91adf037a7e55e09111f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 20:29:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/658775/1/check/openstack-tox-pep8/dc7e4fd/ : SUCCESS in 6m 24s\n- openstack-tox-docs http://logs.openstack.org/75/658775/1/check/openstack-tox-docs/ca36f87/html/ : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":1},{"id":"76510770d7ade4bf4198abb392b1a91f0d8a34e8","author":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"date":"2019-05-16 20:42:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2651d54ad302ecf9ccf3148fc0ddb75a106e3e26","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-05-16 22:16:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, both links LGTM","accounts_in_message":[],"_revision_number":1},{"id":"6265ad0186a40c1fa199515b8d1dbcb3301091aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:17:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46e34dfeadabf3f8ac80a282ca6db0bd0fcf8866","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:50:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/658775/1/gate/openstack-tox-pep8/0978a7c/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/75/658775/1/gate/openstack-tox-docs/df97431/html/ : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5d13f0f9a08df891b7787010de3fb9dbecd39b0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:50:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"788db67b5880dd00be29071cf621cd888bc5f604","revisions":{"788db67b5880dd00be29071cf621cd888bc5f604":{"kind":"REWORK","_number":1,"created":"2019-05-13 09:37:13.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/75/658775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/75/658775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/75/658775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/75/658775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/75/658775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/75/658775/1"}}},"commit":{"parents":[{"commit":"35062ad4152901fb04bb67a9a65dd0b3247ec308","subject":"Dropping the py35 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/35062ad4152901fb04bb67a9a65dd0b3247ec308"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-05-13 09:33:32.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-05-13 09:33:32.000000000","tz":480},"subject":"Fix some inaccurate URL links","message":"Fix some inaccurate URL links\n\nThis is to fix some inaccurate URL links.\n\nChange-Id: Ie43108de22ea749990750f7f20d4d66475d33547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/788db67b5880dd00be29071cf621cd888bc5f604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/788db67b5880dd00be29071cf621cd888bc5f604"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
