)]}'
{"id":"openstack%2Fpatrole~670229","triplet_id":"openstack%2Fpatrole~master~I498c637da9dac7605851b2edbbfb302103c3d6a9","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I498c637da9dac7605851b2edbbfb302103c3d6a9","subject":"Zuul: Fix extra quotes when writing devstack localconf","status":"MERGED","created":"2019-07-11 05:06:57.000000000","updated":"2019-07-17 14:34:20.000000000","submitted":"2019-07-17 14:34:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670229-1563374060790-02c3ebfc","meta_rev_id":"70e55b61f9264768bcb99ecdd2785dd3e4a58e95","_number":670229,"virtual_id_number":670229,"owner":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},{"value":2,"date":"2019-07-17 14:34:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29868,"name":"Mykola Yakovliev","email":"my5878@att.com"},{"value":0,"_account_id":7708,"name":"Sergey Vilgelm","username":"svilgelm","inactive":true},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2019-07-14 04:01:02.000000000","_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2019-07-14 14:36:07.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":1,"date":"2019-07-15 15:02:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29868,"name":"Mykola Yakovliev","email":"my5878@att.com"},{"value":0,"_account_id":7708,"name":"Sergey Vilgelm","username":"svilgelm","inactive":true},{"value":2,"date":"2019-07-17 13:12:51.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29868,"name":"Mykola Yakovliev","email":"my5878@att.com"},{"value":0,"_account_id":7708,"name":"Sergey Vilgelm","username":"svilgelm","inactive":true},{"value":1,"date":"2019-07-17 13:12:51.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7708,"name":"Sergey Vilgelm","username":"svilgelm","inactive":true},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"_account_id":29868,"name":"Mykola Yakovliev","email":"my5878@att.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 05:42:13.000000000","updated_by":{"_account_id":29868,"name":"Mykola Yakovliev","email":"my5878@att.com"},"reviewer":{"_account_id":29868,"name":"Mykola Yakovliev","email":"my5878@att.com"},"state":"REVIEWER"},{"updated":"2019-07-12 05:42:30.000000000","updated_by":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"reviewer":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"state":"REVIEWER"},{"updated":"2019-07-12 16:23:37.000000000","updated_by":{"_account_id":7708,"name":"Sergey Vilgelm","username":"svilgelm","inactive":true},"reviewer":{"_account_id":7708,"name":"Sergey Vilgelm","username":"svilgelm","inactive":true},"state":"REVIEWER"},{"updated":"2019-07-14 14:36:07.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-07-15 15:02:15.000000000","updated_by":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"reviewer":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"state":"REVIEWER"},{"updated":"2019-07-17 13:12:51.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-07-17 14:34:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ebd2356dfb905c1287043df64929b27de179810","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 05:06:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6873d84d3dca9416e4bca26f7b1df7ca900b385b","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 06:08:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"210b1e4455b5d498e2256d5b2e8e006398529d2b","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 07:33:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"11bdb64ef92f2f945da11ebe015943b418f218e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 08:54:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/29/670229/3/check/openstack-tox-cover/3dc0eda/cover/ : SUCCESS in 3m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/670229/3/check/openstack-tox-lower-constraints/71556d6/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/29/670229/3/check/openstack-tox-py36/2f6c5a8/ : SUCCESS in 2m 51s\n- openstack-tox-py37 http://logs.openstack.org/29/670229/3/check/openstack-tox-py37/dd8bddf/ : SUCCESS in 3m 20s\n- openstack-tox-pep8 http://logs.openstack.org/29/670229/3/check/openstack-tox-pep8/f0f829f/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/29/670229/3/check/openstack-tox-py27/4723e98/ : SUCCESS in 2m 35s\n- openstack-tox-docs http://logs.openstack.org/29/670229/3/check/openstack-tox-docs/1fc43ad/html/ : SUCCESS in 2m 24s\n- patrole-admin http://logs.openstack.org/29/670229/3/check/patrole-admin/3babf1c/ : SUCCESS in 54m 03s\n- patrole-member http://logs.openstack.org/29/670229/3/check/patrole-member/ee59582/ : SUCCESS in 1h 14m 18s\n- patrole-reader http://logs.openstack.org/29/670229/3/check/patrole-reader/ae14678/ : SUCCESS in 1h 15m 44s (non-voting)\n- patrole-member-stein http://logs.openstack.org/29/670229/3/check/patrole-member-stein/dcda318/ : FAILURE in 1h 11m 27s\n- patrole-member-rocky http://logs.openstack.org/29/670229/3/check/patrole-member-rocky/e29e63b/ : SUCCESS in 1h 11m 47s\n- patrole-member-queens http://logs.openstack.org/29/670229/3/check/patrole-member-queens/df1e579/ : SUCCESS in 1h 03m 08s\n- patrole-member-pike http://logs.openstack.org/29/670229/3/check/patrole-member-pike/9ffd6d2/ : SUCCESS in 1h 03m 27s\n- patrole-multinode-admin http://logs.openstack.org/29/670229/3/check/patrole-multinode-admin/6a99f88/ : FAILURE in 32m 43s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/29/670229/3/check/patrole-multinode-member/7d13e96/ : FAILURE in 36m 39s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/29/670229/3/check/patrole-extension-admin/7229cea/ : SUCCESS in 34m 00s (non-voting)\n- patrole-extension-member http://logs.openstack.org/29/670229/3/check/patrole-extension-member/e0bd6b8/ : POST_FAILURE in 40m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d479dc2d16fded0e17da858cfcd86b569601aa44","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 14:30:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4720d97ac3aa56cf1ff5eec3720b9545347008d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:12:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/29/670229/4/check/openstack-tox-cover/7189b9d/cover/ : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/670229/4/check/openstack-tox-lower-constraints/653c60e/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/29/670229/4/check/openstack-tox-py36/8cc36cc/ : SUCCESS in 5m 59s\n- openstack-tox-py37 http://logs.openstack.org/29/670229/4/check/openstack-tox-py37/d2815df/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/29/670229/4/check/openstack-tox-pep8/ca156ea/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/29/670229/4/check/openstack-tox-py27/e309dcd/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/29/670229/4/check/openstack-tox-docs/f7d0706/html/ : SUCCESS in 5m 50s\n- patrole-admin http://logs.openstack.org/29/670229/4/check/patrole-admin/7825c95/ : SUCCESS in 56m 03s\n- patrole-member http://logs.openstack.org/29/670229/4/check/patrole-member/4caeabf/ : SUCCESS in 1h 17m 03s\n- patrole-reader http://logs.openstack.org/29/670229/4/check/patrole-reader/9c949b9/ : SUCCESS in 1h 23m 24s (non-voting)\n- patrole-member-stein http://logs.openstack.org/29/670229/4/check/patrole-member-stein/239e2c7/ : SUCCESS in 1h 08m 00s\n- patrole-member-rocky http://logs.openstack.org/29/670229/4/check/patrole-member-rocky/052ffbd/ : SUCCESS in 1h 14m 41s\n- patrole-member-queens http://logs.openstack.org/29/670229/4/check/patrole-member-queens/2e2f7cf/ : SUCCESS in 1h 13m 19s\n- patrole-member-pike http://logs.openstack.org/29/670229/4/check/patrole-member-pike/67619e1/ : POST_FAILURE in 1h 14m 09s\n- patrole-multinode-admin http://logs.openstack.org/29/670229/4/check/patrole-multinode-admin/8465d52/ : FAILURE in 33m 09s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/29/670229/4/check/patrole-multinode-member/3277858/ : FAILURE in 42m 05s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/29/670229/4/check/patrole-extension-admin/da1b47f/ : SUCCESS in 35m 01s (non-voting)\n- patrole-extension-member http://logs.openstack.org/29/670229/4/check/patrole-extension-member/9206529/ : FAILURE in 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"79199b6dd95146011f357a7ccc62f7cfc734415b","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 16:13:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae527b0acf9e8e85c3ab17dceff015bce04bdfee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:13:36.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/patrole on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/patrole/.zuul.yaml@master\", line 56, column 9\nexpected \u003cblock end\u003e, but found \u0027\u003cscalar\u003e\u0027\n  in \"openstack/patrole/.zuul.yaml@master\", line 56, column 48","accounts_in_message":[],"_revision_number":5},{"id":"7f2f4b4460400fcd8ba6955b7abb635f492d20b8","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 16:19:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0eeef8ce1a651c092d835e6860e05ab150c38de3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:20:58.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/patrole on branch master.  The error was:\n\n  Invalid Ansible variable name \u0027devstack_localrc:\u0027 for dictionary value @ data[\u0027vars\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: patrole-extension-base\n      parent: patrole-base\n      description: |\n        Patrole plugin job for admin and member roles which runs RBAC tests for\n        neutron-tempest-plugin APIs (if the plugin is installed).\n  \n        Covers Neutron extension functionality only. Should not be used for\n        supporting Neutron plugins like fwaas.\n      required-projects:\n        - name: openstack/tempest\n        - name: openstack/patrole\n        - name: openstack/neutron-tempest-plugin\n      vars:\n        devstack_localrc::\n          TEMPEST_PLUGINS: \u0027\"/opt/stack/patrole /opt/stack/neutron-tempest-plugin\"\u0027\n        devstack_plugins:\n          neutron: https://opendev.org/openstack/neutron.git\n          patrole: https://opendev.org/openstack/patrole.git\n          neutron-tempest-plugin: https://opendev.org/openstack/neutron-tempest-plugin.git\n        devstack_services:\n          tempest: true\n          neutron: true\n          neutron-segments: true\n          neutron-qos: true\n        tempest_test_regex: (?\u003d.*ExtRbacTest)(^patrole_tempest_plugin\\.tests\\.api)\n\n  in \"openstack/patrole/.zuul.yaml@master\", line 153, column 3","accounts_in_message":[],"_revision_number":6},{"id":"bfe2d8191ae1af41c051ddade19ffa244b126574","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 16:29:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8bacc4363437280b0ea62cc8d6fa29e026975f61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 17:51:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/29/670229/7/check/openstack-tox-cover/b1e9787/cover/ : SUCCESS in 2m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/670229/7/check/openstack-tox-lower-constraints/1d54de0/ : SUCCESS in 2m 54s\n- openstack-tox-py36 http://logs.openstack.org/29/670229/7/check/openstack-tox-py36/2e17fa6/ : SUCCESS in 3m 04s\n- openstack-tox-py37 http://logs.openstack.org/29/670229/7/check/openstack-tox-py37/726338a/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/29/670229/7/check/openstack-tox-pep8/2702729/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/29/670229/7/check/openstack-tox-py27/c77cd08/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/29/670229/7/check/openstack-tox-docs/2dc4081/html/ : SUCCESS in 2m 58s\n- patrole-admin http://logs.openstack.org/29/670229/7/check/patrole-admin/31496a8/ : SUCCESS in 1h 02m 51s\n- patrole-member http://logs.openstack.org/29/670229/7/check/patrole-member/98f4fb1/ : SUCCESS in 1h 14m 37s\n- patrole-reader http://logs.openstack.org/29/670229/7/check/patrole-reader/c28edf4/ : SUCCESS in 1h 14m 47s (non-voting)\n- patrole-member-stein http://logs.openstack.org/29/670229/7/check/patrole-member-stein/3fb7c41/ : SUCCESS in 1h 03m 15s\n- patrole-member-rocky http://logs.openstack.org/29/670229/7/check/patrole-member-rocky/e71c75f/ : SUCCESS in 1h 11m 16s\n- patrole-member-queens http://logs.openstack.org/29/670229/7/check/patrole-member-queens/e3bf23e/ : SUCCESS in 1h 07m 06s\n- patrole-member-pike http://logs.openstack.org/29/670229/7/check/patrole-member-pike/70badbb/ : SUCCESS in 1h 14m 25s\n- patrole-multinode-admin http://logs.openstack.org/29/670229/7/check/patrole-multinode-admin/e72fcab/ : FAILURE in 38m 39s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/29/670229/7/check/patrole-multinode-member/32869a6/ : FAILURE in 35m 11s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/29/670229/7/check/patrole-extension-admin/ecd4e1e/ : SUCCESS in 36m 51s (non-voting)\n- patrole-extension-member http://logs.openstack.org/29/670229/7/check/patrole-extension-member/7f8f418/ : FAILURE in 42m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c5a8e836e52ae9f4aca921cf776100dc488730d4","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 17:58:40.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"e97171d40cd645744b8bbdb3db9a105566009048","author":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"date":"2019-07-11 18:22:09.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"cb4001d84b33b9caa33b22981962976e7e95bbab","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 21:30:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"423781f42e824729e8a2195f6c287089bac46545","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 22:17:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2e0c970d584ef0269400625b5489b9edde78a4d0","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-11 23:55:18.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4693b5883201b65fe40bae4cb79bf1e19fce7be4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 01:13:09.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/29/670229/10/check/openstack-tox-cover/e6b997e/cover/ : SUCCESS in 2m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/670229/10/check/openstack-tox-lower-constraints/2e80f4b/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/29/670229/10/check/openstack-tox-py36/4e6fa87/ : SUCCESS in 2m 48s\n- openstack-tox-py37 http://logs.openstack.org/29/670229/10/check/openstack-tox-py37/d81c713/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/29/670229/10/check/openstack-tox-pep8/dd4e16c/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/29/670229/10/check/openstack-tox-py27/7166213/ : SUCCESS in 2m 56s\n- openstack-tox-docs http://logs.openstack.org/29/670229/10/check/openstack-tox-docs/5cfe1dd/html/ : SUCCESS in 2m 46s\n- patrole-admin http://logs.openstack.org/29/670229/10/check/patrole-admin/25a90d7/ : SUCCESS in 58m 55s\n- patrole-member http://logs.openstack.org/29/670229/10/check/patrole-member/f130d31/ : SUCCESS in 1h 06m 30s\n- patrole-reader http://logs.openstack.org/29/670229/10/check/patrole-reader/3ddf488/ : SUCCESS in 1h 04m 20s (non-voting)\n- patrole-member-stein http://logs.openstack.org/29/670229/10/check/patrole-member-stein/1951b27/ : FAILURE in 1h 00m 58s\n- patrole-member-rocky http://logs.openstack.org/29/670229/10/check/patrole-member-rocky/8364b5d/ : SUCCESS in 1h 06m 51s\n- patrole-member-queens http://logs.openstack.org/29/670229/10/check/patrole-member-queens/48ffb0b/ : SUCCESS in 1h 03m 58s\n- patrole-member-pike http://logs.openstack.org/29/670229/10/check/patrole-member-pike/689e9e9/ : SUCCESS in 1h 02m 41s\n- patrole-multinode-admin http://logs.openstack.org/29/670229/10/check/patrole-multinode-admin/f76097f/ : FAILURE in 31m 04s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/29/670229/10/check/patrole-multinode-member/0166acb/ : FAILURE in 31m 21s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/29/670229/10/check/patrole-extension-admin/f1d563f/ : SUCCESS in 30m 57s (non-voting)\n- patrole-extension-member http://logs.openstack.org/29/670229/10/check/patrole-extension-member/879b963/ : FAILURE in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"529376e270c3bc3f76711661ee89c177175ab800","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-12 04:20:53.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"a66fbe2686688659a32822691256c1a48cd20a81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 05:41:14.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/29/670229/10/check/openstack-tox-cover/18bd27c/cover/ : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/670229/10/check/openstack-tox-lower-constraints/b7aa54a/ : SUCCESS in 2m 36s\n- openstack-tox-py36 http://logs.openstack.org/29/670229/10/check/openstack-tox-py36/b863137/ : SUCCESS in 2m 40s\n- openstack-tox-py37 http://logs.openstack.org/29/670229/10/check/openstack-tox-py37/65f07ba/ : SUCCESS in 2m 59s\n- openstack-tox-pep8 http://logs.openstack.org/29/670229/10/check/openstack-tox-pep8/e437fa7/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/29/670229/10/check/openstack-tox-py27/e0cbcff/ : SUCCESS in 2m 38s\n- openstack-tox-docs http://logs.openstack.org/29/670229/10/check/openstack-tox-docs/c47445b/html/ : SUCCESS in 2m 18s\n- patrole-admin http://logs.openstack.org/29/670229/10/check/patrole-admin/a7c5ff5/ : SUCCESS in 58m 02s\n- patrole-member http://logs.openstack.org/29/670229/10/check/patrole-member/ebf326c/ : SUCCESS in 1h 08m 24s\n- patrole-reader http://logs.openstack.org/29/670229/10/check/patrole-reader/816affc/ : SUCCESS in 1h 03m 33s (non-voting)\n- patrole-member-stein http://logs.openstack.org/29/670229/10/check/patrole-member-stein/996921b/ : SUCCESS in 1h 00m 47s\n- patrole-member-rocky http://logs.openstack.org/29/670229/10/check/patrole-member-rocky/87052ea/ : SUCCESS in 1h 19m 47s\n- patrole-member-queens http://logs.openstack.org/29/670229/10/check/patrole-member-queens/bbeda6a/ : SUCCESS in 1h 07m 09s\n- patrole-member-pike http://logs.openstack.org/29/670229/10/check/patrole-member-pike/71491e2/ : SUCCESS in 1h 09m 47s\n- patrole-multinode-admin http://logs.openstack.org/29/670229/10/check/patrole-multinode-admin/d8f5738/ : FAILURE in 34m 45s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/29/670229/10/check/patrole-multinode-member/2ae2950/ : FAILURE in 35m 23s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/29/670229/10/check/patrole-extension-admin/5ddc0f8/ : SUCCESS in 32m 31s (non-voting)\n- patrole-extension-member http://logs.openstack.org/29/670229/10/check/patrole-extension-member/65aa552/ : FAILURE in 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f4c3c211179d7bd7905e850c4df697e978998d8c","author":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"date":"2019-07-12 16:23:23.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"56d15bcdf71e278cd87bf50d649b544b5355b733","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-13 11:58:46.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0dfaa169160017962e1a111e51f89686274a9bb2","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-13 18:12:51.000000000","message":"Patch Set 10:\n\n(2 comments)\n\nAgreed, but I don\u0027t think there is a great way to do this without restructuring the inheritance of the jobs, creating new ones and renaming others. I can try that, but this seems to be the minimal set of changes to get things running again.","accounts_in_message":[],"_revision_number":10},{"id":"274c8b74804490d91bc43b753d70932ce19bdf29","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-13 23:28:21.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"452967470032fa3961b5fa0eb8125b43022dce39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 00:48:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/29/670229/11/check/openstack-tox-cover/8278d27/cover/ : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/670229/11/check/openstack-tox-lower-constraints/c87230e/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/29/670229/11/check/openstack-tox-py36/360f694/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/29/670229/11/check/openstack-tox-py37/790e1f3/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/29/670229/11/check/openstack-tox-pep8/b58b33d/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/29/670229/11/check/openstack-tox-py27/a2120aa/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/29/670229/11/check/openstack-tox-docs/a2d78bf/html/ : SUCCESS in 4m 41s\n- patrole-admin http://logs.openstack.org/29/670229/11/check/patrole-admin/23a6dca/ : SUCCESS in 50m 39s\n- patrole-member http://logs.openstack.org/29/670229/11/check/patrole-member/608e59d/ : SUCCESS in 1h 05m 05s\n- patrole-reader http://logs.openstack.org/29/670229/11/check/patrole-reader/cf3c3d9/ : SUCCESS in 1h 06m 20s (non-voting)\n- patrole-member-stein http://logs.openstack.org/29/670229/11/check/patrole-member-stein/ffdc1d3/ : SUCCESS in 1h 07m 48s\n- patrole-member-rocky http://logs.openstack.org/29/670229/11/check/patrole-member-rocky/a8e6d8d/ : SUCCESS in 1h 04m 52s\n- patrole-member-queens http://logs.openstack.org/29/670229/11/check/patrole-member-queens/72ebb10/ : SUCCESS in 1h 06m 00s\n- patrole-member-pike http://logs.openstack.org/29/670229/11/check/patrole-member-pike/9a51604/ : SUCCESS in 1h 06m 45s\n- patrole-multinode-admin http://logs.openstack.org/29/670229/11/check/patrole-multinode-admin/bd3c9d4/ : FAILURE in 31m 02s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/29/670229/11/check/patrole-multinode-member/aae8b91/ : FAILURE in 30m 34s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/29/670229/11/check/patrole-extension-admin/a7be1d5/ : SUCCESS in 33m 19s (non-voting)\n- patrole-extension-member http://logs.openstack.org/29/670229/11/check/patrole-extension-member/6037693/ : FAILURE in 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"898eb3b13990c16e756aa8f9032e15c242b29cbf","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-14 01:19:17.000000000","message":"Patch Set 11:\n\nThe multinode jobs are failing because they are inheriting from the .zuul.yaml in the stable/stein patrole branch. Per the Zuul docs [0] this is expected. Depending on what the plan is for keeping stable/stein, that .zuul.yaml may also need to be updated. It may also make sense to remove the `branches:` specifications from some of the jobs.\n\nInheritance path from inventory.yaml [1]:\n\n      - \u0027\u003cJob patrole-base-multinode branches: {MatchAny:{BranchMatcher:master}} source:\n        openstack/patrole/.zuul.yaml@master#37\u003e\u0027\n      - \u0027\u003cJob patrole-base-multinode branches: {MatchAny:{BranchMatcher:master}} source:\n        openstack/patrole/.zuul.yaml@stable/stein#37\u003e\u0027\n      - \u0027\u003cJob patrole-multinode-admin branches: {MatchAny:{BranchMatcher:master}}\n        source: openstack/patrole/.zuul.yaml@master#128\u003e\u0027\n      - \u0027\u003cJob patrole-multinode-admin branches: None source: openstack/patrole/.zuul.yaml@master#206\u003e\u0027\n\n[0] https://zuul-ci.org/docs/zuul/user/config.html#configuration-loading\n[1] http://logs.openstack.org/29/670229/11/check/patrole-multinode-admin/bd3c9d4/zuul-info/inventory.yaml","accounts_in_message":[],"_revision_number":11},{"id":"0413f9947b1d0b5544a0573542884f0e7563f062","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2019-07-14 04:01:02.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"98bf639381431d273eff6f2c693d3bf96dc3b0e6","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-07-14 14:36:07.000000000","message":"Patch Set 11: Code-Review+2\n\nThanks for doing this.","accounts_in_message":[],"_revision_number":11},{"id":"85fc5b0d42defdeb31f6c4138fdc81f11eff1982","author":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"date":"2019-07-15 15:02:15.000000000","message":"Patch Set 11: Code-Review+1\n\nThanks Phil!","accounts_in_message":[],"_revision_number":11},{"id":"2a796252fef82f25a891b0d6afc48184f55d44b3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-17 13:12:51.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"ce1383222190b879f7ac450353e73f02a4df9596","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 13:13:08.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"6aa7e49dc53245c3a01db6b76bcc73e4a7eb4f13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:34:20.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/29/670229/11/gate/openstack-tox-lower-constraints/a4eea3e/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/29/670229/11/gate/openstack-tox-py36/52850f5/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/29/670229/11/gate/openstack-tox-py37/2fb7998/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 http://logs.openstack.org/29/670229/11/gate/openstack-tox-pep8/eac700f/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/29/670229/11/gate/openstack-tox-py27/0e09395/ : SUCCESS in 2m 57s\n- openstack-tox-docs http://logs.openstack.org/29/670229/11/gate/openstack-tox-docs/f6b04dd/html/ : SUCCESS in 3m 11s\n- patrole-admin http://logs.openstack.org/29/670229/11/gate/patrole-admin/52967d0/ : SUCCESS in 58m 09s\n- patrole-member http://logs.openstack.org/29/670229/11/gate/patrole-member/ffce32c/ : SUCCESS in 1h 14m 55s","accounts_in_message":[],"_revision_number":11},{"id":"70e55b61f9264768bcb99ecdd2785dd3e4a58e95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:34:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"6b229a720e88f7385f027c9c5307d4f115ba95f3","revisions":{"6f9f96aeccecc9965d24c79189ef4a2f240250d8":{"kind":"REWORK","_number":1,"created":"2019-07-11 05:06:57.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/1"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf","message":"Zuul: Fix extra quotes when writing devstack localconf\n\nWhen [0] introduced quoting all arguments, it broke the Zuul jobs\nbecause the value of TEMPEST_PLUGINS is quoted. The fix [1] only\naddressed extra double-quotes, not extra single-quotes.\n\nThis patchset changes from single to double quotes, which should\nprovide compatibility with all releases.\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6f9f96aeccecc9965d24c79189ef4a2f240250d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6f9f96aeccecc9965d24c79189ef4a2f240250d8"}]},"branch":"refs/heads/master"},"ea08a46bd4eeaabd9b487f8620efdf8dcdafc5e0":{"kind":"REWORK","_number":2,"created":"2019-07-11 06:08:17.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/2"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 06:07:03.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf","message":"Zuul: Fix extra quotes when writing devstack localconf\n\nWhen [0] introduced quoting all arguments, it broke the Zuul jobs\nbecause the value of TEMPEST_PLUGINS is quoted. The fix [1] only\naddressed extra double-quotes, not extra single-quotes.\n\nThis patchset changes from single to double quotes where required,\nand removes extraneous quotes, which should provide compatibility\nwith all releases.\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ea08a46bd4eeaabd9b487f8620efdf8dcdafc5e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ea08a46bd4eeaabd9b487f8620efdf8dcdafc5e0"}]},"branch":"refs/heads/master"},"732a8162b97deeb1b23871553620d4a34a9b262d":{"kind":"REWORK","_number":3,"created":"2019-07-11 07:33:17.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/3"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 07:32:37.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf","message":"Zuul: Fix extra quotes when writing devstack localconf\n\nWhen [0] introduced quoting all arguments, it broke the Zuul jobs\nbecause the value of TEMPEST_PLUGINS is quoted. The fix [1] only\naddressed extra double-quotes, not extra single-quotes.\n\nThis patchset changes from single to double quotes where required,\nand removes extraneous quotes, which should provide compatibility\nwith all releases.\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/732a8162b97deeb1b23871553620d4a34a9b262d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/732a8162b97deeb1b23871553620d4a34a9b262d"}]},"branch":"refs/heads/master"},"2c99a11677f56bf704898d904e3501b3afe30ce9":{"kind":"REWORK","_number":4,"created":"2019-07-11 14:30:10.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/4"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 14:26:08.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf (WIP)","message":"Zuul: Fix extra quotes when writing devstack localconf (WIP)\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n[2] https://review.opendev.org/639374\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2c99a11677f56bf704898d904e3501b3afe30ce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2c99a11677f56bf704898d904e3501b3afe30ce9"}]},"branch":"refs/heads/master"},"2156065423b224a2ef6bdd71679e79880ca2c2bf":{"kind":"REWORK","_number":5,"created":"2019-07-11 16:13:23.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/5"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"Sphicas, Phil (ps3910)","email":"ps3910@att.com","date":"2019-07-11 16:12:49.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf (WIP)","message":"Zuul: Fix extra quotes when writing devstack localconf (WIP)\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n[2] https://review.opendev.org/639374\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2156065423b224a2ef6bdd71679e79880ca2c2bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2156065423b224a2ef6bdd71679e79880ca2c2bf"}]},"branch":"refs/heads/master"},"d0df26e4ba7b57c11da2499fc5bcec37f59cf43e":{"kind":"REWORK","_number":6,"created":"2019-07-11 16:19:04.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/6"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"Sphicas, Phil (ps3910)","email":"ps3910@att.com","date":"2019-07-11 16:18:57.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf (WIP)","message":"Zuul: Fix extra quotes when writing devstack localconf (WIP)\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n[2] https://review.opendev.org/639374\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d0df26e4ba7b57c11da2499fc5bcec37f59cf43e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/d0df26e4ba7b57c11da2499fc5bcec37f59cf43e"}]},"branch":"refs/heads/master"},"38cef3cd8ca9a0d4f2f0e8daeb15a731309d1239":{"kind":"REWORK","_number":7,"created":"2019-07-11 16:29:12.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/7"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"Sphicas, Phil (ps3910)","email":"ps3910@att.com","date":"2019-07-11 16:29:04.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf (WIP)","message":"Zuul: Fix extra quotes when writing devstack localconf (WIP)\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n[2] https://review.opendev.org/639374\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/38cef3cd8ca9a0d4f2f0e8daeb15a731309d1239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/38cef3cd8ca9a0d4f2f0e8daeb15a731309d1239"}]},"branch":"refs/heads/master"},"4947f0b1b7facba7024e75d46895b2c69fd63ff1":{"kind":"REWORK","_number":8,"created":"2019-07-11 21:30:02.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/8"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"Sphicas, Phil (ps3910)","email":"ps3910@att.com","date":"2019-07-11 21:22:58.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf","message":"Zuul: Fix extra quotes when writing devstack localconf\n\nGenerate the correct TEMPEST_PLUGINS environment variable, working\naround devstack\u0027s behavior of quoting all arguments [0]. The fix [1]\nonly addressed extra double-quotes, not extra single-quotes.\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4947f0b1b7facba7024e75d46895b2c69fd63ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4947f0b1b7facba7024e75d46895b2c69fd63ff1"}]},"branch":"refs/heads/master"},"4fd359fae14de204eb70ddaae8752edeea9786da":{"kind":"REWORK","_number":9,"created":"2019-07-11 22:17:09.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/9"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"Sphicas, Phil (ps3910)","email":"ps3910@att.com","date":"2019-07-11 22:16:48.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf","message":"Zuul: Fix extra quotes when writing devstack localconf\n\nGenerate the correct TEMPEST_PLUGINS environment variable, working\naround devstack\u0027s behavior of quoting all arguments [0]. The fix [1]\nonly addressed extra double-quotes, not extra single-quotes.\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4fd359fae14de204eb70ddaae8752edeea9786da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4fd359fae14de204eb70ddaae8752edeea9786da"}]},"branch":"refs/heads/master"},"f0d7a5e23fc2be5ef7641b6fdfe7bb084eedcc36":{"kind":"REWORK","_number":10,"created":"2019-07-11 23:55:18.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/10"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"Sphicas, Phil (ps3910)","email":"ps3910@att.com","date":"2019-07-11 23:44:31.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf","message":"Zuul: Fix extra quotes when writing devstack localconf\n\nGenerate the correct TEMPEST_PLUGINS environment variable, working\naround devstack\u0027s behavior of quoting all arguments [0]. For the\npatrole-base and patrole-extension-base jobs, it is sufficient to\nsimply remove the quotes. For the multinode job, there is no way to\navoid quotes, and [1] is not able to remove extra single-quotes, so\nwe move to the new variable tempest_plugins introduced with [2].\n\n[0] https://review.opendev.org/636078\n[1] https://review.opendev.org/648951\n[2] https://review.opendev.org/639374\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f0d7a5e23fc2be5ef7641b6fdfe7bb084eedcc36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f0d7a5e23fc2be5ef7641b6fdfe7bb084eedcc36"}]},"branch":"refs/heads/master"},"6b229a720e88f7385f027c9c5307d4f115ba95f3":{"kind":"REWORK","_number":11,"created":"2019-07-13 23:28:21.000000000","uploader":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"ref":"refs/changes/29/670229/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/29/670229/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/29/670229/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/29/670229/11"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"ps3910","email":"ps3910@att.com","date":"2019-07-11 04:54:44.000000000","tz":-420},"committer":{"name":"Sphicas, Phil (ps3910)","email":"ps3910@att.com","date":"2019-07-13 23:24:27.000000000","tz":-420},"subject":"Zuul: Fix extra quotes when writing devstack localconf","message":"Zuul: Fix extra quotes when writing devstack localconf\n\nGenerate the correct TEMPEST_PLUGINS environment variable, moving to\nthe new variable tempest_plugins introduced with [0], to get around\ndevstack\u0027s behavior of quoting all arguments [1]. Maintain the old\nsyntax for releases earlier than stein.\n\n[0] https://review.opendev.org/639374\n[1] https://review.opendev.org/636078\n\nChange-Id: I498c637da9dac7605851b2edbbfb302103c3d6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6b229a720e88f7385f027c9c5307d4f115ba95f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6b229a720e88f7385f027c9c5307d4f115ba95f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
