)]}'
{"id":"openstack%2Fpatrole~670637","triplet_id":"openstack%2Fpatrole~master~Ib3e0b4220d9a8a076ba2c26ac68852aa3953f134","project":"openstack/patrole","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ib3e0b4220d9a8a076ba2c26ac68852aa3953f134","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-12 20:08:04.000000000","updated":"2019-07-18 21:39:31.000000000","submitted":"2019-07-18 21:39:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670637-1563485971366-0ae53db0","meta_rev_id":"6fadc42ab1472065dfbd2d67fae7e91055f752c5","_number":670637,"virtual_id_number":670637,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-07-18 21:39:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2019-07-17 02:11:53.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2019-07-16 04:27:46.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":1,"date":"2019-07-18 18:21:28.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 02:11:53.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-07-18 18:21:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-07-18 21:39:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9ab715bf57191a0dcdadb8671584364e3480574","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-12 20:08:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f01b817c05238c2e851cb66a422de8b4b1461a2a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-12 20:10:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b7d5d31fc171eb5923e602cead5a70b53b050de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 21:30:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/37/670637/1/check/test-release-openstack/39b5bd0/ : SUCCESS in 2m 46s\n- openstack-tox-cover http://logs.openstack.org/37/670637/1/check/openstack-tox-cover/caad0c0/cover/ : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/670637/1/check/openstack-tox-lower-constraints/b5b9592/ : SUCCESS in 3m 40s\n- openstack-tox-py36 http://logs.openstack.org/37/670637/1/check/openstack-tox-py36/5728a22/ : SUCCESS in 3m 34s\n- openstack-tox-py37 http://logs.openstack.org/37/670637/1/check/openstack-tox-py37/4dc0ef3/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/37/670637/1/check/openstack-tox-pep8/ef2c13a/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/37/670637/1/check/openstack-tox-py27/b020f02/ : SUCCESS in 3m 44s\n- openstack-tox-docs http://logs.openstack.org/37/670637/1/check/openstack-tox-docs/f26dd81/html/ : SUCCESS in 2m 56s\n- build-openstack-releasenotes http://logs.openstack.org/37/670637/1/check/build-openstack-releasenotes/b3b9ca9/html/ : SUCCESS in 2m 58s\n- patrole-admin http://logs.openstack.org/37/670637/1/check/patrole-admin/7b46572/ : FAILURE in 32m 17s\n- patrole-member http://logs.openstack.org/37/670637/1/check/patrole-member/aa0fc42/ : FAILURE in 31m 35s\n- patrole-reader http://logs.openstack.org/37/670637/1/check/patrole-reader/0cd96e5/ : FAILURE in 36m 45s (non-voting)\n- patrole-member-stein http://logs.openstack.org/37/670637/1/check/patrole-member-stein/55952f8/ : SUCCESS in 1h 09m 53s\n- patrole-member-rocky http://logs.openstack.org/37/670637/1/check/patrole-member-rocky/7407bc6/ : SUCCESS in 1h 14m 51s\n- patrole-member-queens http://logs.openstack.org/37/670637/1/check/patrole-member-queens/1f0500c/ : SUCCESS in 1h 02m 41s\n- patrole-member-pike http://logs.openstack.org/37/670637/1/check/patrole-member-pike/2acb713/ : SUCCESS in 1h 15m 39s\n- patrole-multinode-admin http://logs.openstack.org/37/670637/1/check/patrole-multinode-admin/b452239/ : FAILURE in 34m 53s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/37/670637/1/check/patrole-multinode-member/f5497fc/ : FAILURE in 41m 02s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/37/670637/1/check/patrole-extension-admin/821d4ce/ : FAILURE in 31m 10s (non-voting)\n- patrole-extension-member http://logs.openstack.org/37/670637/1/check/patrole-extension-member/c515071/ : FAILURE in 36m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93c41585c2848354efa16950030adc8230e01462","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-15 17:19:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4cd4565b0cd934f8dd80636005d3fdd05c8aabcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 18:45:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/37/670637/2/check/test-release-openstack/ef8a197/ : SUCCESS in 2m 20s\n- openstack-tox-cover http://logs.openstack.org/37/670637/2/check/openstack-tox-cover/6423f47/cover/ : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/670637/2/check/openstack-tox-lower-constraints/1209daa/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/37/670637/2/check/openstack-tox-py36/6f287fc/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/37/670637/2/check/openstack-tox-py37/811033e/ : SUCCESS in 3m 11s\n- openstack-tox-pep8 http://logs.openstack.org/37/670637/2/check/openstack-tox-pep8/51d3987/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/37/670637/2/check/openstack-tox-py27/4859e21/ : SUCCESS in 2m 51s\n- openstack-tox-docs http://logs.openstack.org/37/670637/2/check/openstack-tox-docs/08a0592/html/ : SUCCESS in 4m 13s\n- patrole-admin http://logs.openstack.org/37/670637/2/check/patrole-admin/ff9ed47/ : FAILURE in 35m 01s\n- patrole-member http://logs.openstack.org/37/670637/2/check/patrole-member/05c9cd1/ : FAILURE in 34m 17s\n- patrole-reader http://logs.openstack.org/37/670637/2/check/patrole-reader/cb4e1b5/ : FAILURE in 37m 22s (non-voting)\n- patrole-member-stein http://logs.openstack.org/37/670637/2/check/patrole-member-stein/baa8152/ : FAILURE in 1h 10m 16s\n- patrole-member-rocky http://logs.openstack.org/37/670637/2/check/patrole-member-rocky/3907a00/ : SUCCESS in 1h 17m 53s\n- patrole-member-queens http://logs.openstack.org/37/670637/2/check/patrole-member-queens/51601ed/ : SUCCESS in 1h 14m 59s\n- patrole-member-pike http://logs.openstack.org/37/670637/2/check/patrole-member-pike/0a46e19/ : SUCCESS in 1h 09m 21s\n- patrole-multinode-admin http://logs.openstack.org/37/670637/2/check/patrole-multinode-admin/ecbdbc6/ : FAILURE in 34m 57s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/37/670637/2/check/patrole-multinode-member/372ae3e/ : FAILURE in 41m 17s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/37/670637/2/check/patrole-extension-admin/b86294e/ : FAILURE in 39m 12s (non-voting)\n- patrole-extension-member http://logs.openstack.org/37/670637/2/check/patrole-extension-member/6329d75/ : FAILURE in 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b429043f49dfedcb6feb2aa04e534d0b58663ec","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-16 04:27:46.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"56103a4232d87f36d4392df45dcbbff242d62a6e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-16 04:27:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0c65485ac086570241f389bd87fbe0c77b7c9473","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 06:08:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/37/670637/2/check/test-release-openstack/bf37bf6/ : SUCCESS in 2m 00s\n- openstack-tox-cover http://logs.openstack.org/37/670637/2/check/openstack-tox-cover/505db48/cover/ : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/670637/2/check/openstack-tox-lower-constraints/dcefabd/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/37/670637/2/check/openstack-tox-py36/38d637b/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/37/670637/2/check/openstack-tox-py37/efe7728/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/37/670637/2/check/openstack-tox-pep8/ea446fb/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/37/670637/2/check/openstack-tox-py27/5a043f2/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/37/670637/2/check/openstack-tox-docs/9cdf26c/html/ : SUCCESS in 3m 47s\n- patrole-admin http://logs.openstack.org/37/670637/2/check/patrole-admin/d1bdfa7/ : FAILURE in 40m 49s\n- patrole-member http://logs.openstack.org/37/670637/2/check/patrole-member/f785987/ : FAILURE in 31m 09s\n- patrole-reader http://logs.openstack.org/37/670637/2/check/patrole-reader/73a90b2/ : FAILURE in 42m 28s (non-voting)\n- patrole-member-stein http://logs.openstack.org/37/670637/2/check/patrole-member-stein/a34f609/ : SUCCESS in 1h 33m 02s\n- patrole-member-rocky http://logs.openstack.org/37/670637/2/check/patrole-member-rocky/b288b3a/ : SUCCESS in 1h 39m 42s\n- patrole-member-queens http://logs.openstack.org/37/670637/2/check/patrole-member-queens/620f2b3/ : SUCCESS in 1h 26m 09s\n- patrole-member-pike http://logs.openstack.org/37/670637/2/check/patrole-member-pike/ec151b1/ : SUCCESS in 1h 27m 33s\n- patrole-multinode-admin http://logs.openstack.org/37/670637/2/check/patrole-multinode-admin/7c83df1/ : FAILURE in 47m 36s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/37/670637/2/check/patrole-multinode-member/b04fc56/ : FAILURE in 30m 49s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/37/670637/2/check/patrole-extension-admin/fd1fed5/ : FAILURE in 28m 48s (non-voting)\n- patrole-extension-member http://logs.openstack.org/37/670637/2/check/patrole-extension-member/5dc1295/ : FAILURE in 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e09300d3088acf7588608438c7ce66fb556998e","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-07-17 02:11:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6af819cdb9a0457aa2fb0bf967d167648a85f2cb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-18 18:21:28.000000000","message":"Patch Set 2: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e5ee80f6d5e157368a48da264c5dd91df0409a44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 20:27:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/37/670637/2/check/test-release-openstack/de3bfb3/ : SUCCESS in 2m 45s\n- openstack-tox-cover http://logs.openstack.org/37/670637/2/check/openstack-tox-cover/13ae3dc/cover/ : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/670637/2/check/openstack-tox-lower-constraints/b412356/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/37/670637/2/check/openstack-tox-py36/f06fa2e/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/37/670637/2/check/openstack-tox-py37/af747d0/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/37/670637/2/check/openstack-tox-pep8/a2e2b27/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/37/670637/2/check/openstack-tox-py27/e487cc0/ : SUCCESS in 3m 38s\n- openstack-tox-docs http://logs.openstack.org/37/670637/2/check/openstack-tox-docs/e180e7d/html/ : SUCCESS in 3m 42s\n- patrole-admin http://logs.openstack.org/37/670637/2/check/patrole-admin/0a6195f/ : SUCCESS in 58m 22s\n- patrole-member http://logs.openstack.org/37/670637/2/check/patrole-member/f565f73/ : SUCCESS in 1h 14m 12s\n- patrole-reader http://logs.openstack.org/37/670637/2/check/patrole-reader/ae8039b/ : SUCCESS in 1h 17m 33s (non-voting)\n- patrole-member-stein http://logs.openstack.org/37/670637/2/check/patrole-member-stein/cf5b53e/ : SUCCESS in 1h 17m 51s\n- patrole-member-rocky http://logs.openstack.org/37/670637/2/check/patrole-member-rocky/4670dd5/ : SUCCESS in 1h 21m 42s\n- patrole-member-queens http://logs.openstack.org/37/670637/2/check/patrole-member-queens/34e63fd/ : SUCCESS in 1h 54m 23s\n- patrole-member-pike http://logs.openstack.org/37/670637/2/check/patrole-member-pike/d789440/ : SUCCESS in 1h 05m 07s\n- patrole-multinode-admin http://logs.openstack.org/37/670637/2/check/patrole-multinode-admin/4754a13/ : FAILURE in 34m 29s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/37/670637/2/check/patrole-multinode-member/82789fd/ : FAILURE in 36m 40s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/37/670637/2/check/patrole-extension-admin/4477e06/ : SUCCESS in 36m 11s (non-voting)\n- patrole-extension-member http://logs.openstack.org/37/670637/2/check/patrole-extension-member/8b79d75/ : FAILURE in 36m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"078cfc083a22792b46d67b3b03d45c452729b3e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 20:27:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"48f7ca3ff6f6dc0365ac05af19d8aeabd2eeba5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 21:39:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/37/670637/2/gate/test-release-openstack/7ba3d6f/ : SUCCESS in 2m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/670637/2/gate/openstack-tox-lower-constraints/fc70ff7/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/37/670637/2/gate/openstack-tox-py36/c9fd605/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/37/670637/2/gate/openstack-tox-py37/0cc8b3f/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/37/670637/2/gate/openstack-tox-pep8/22afa25/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/37/670637/2/gate/openstack-tox-py27/daf10c7/ : SUCCESS in 3m 05s\n- openstack-tox-docs http://logs.openstack.org/37/670637/2/gate/openstack-tox-docs/a1da023/html/ : SUCCESS in 2m 36s\n- patrole-admin http://logs.openstack.org/37/670637/2/gate/patrole-admin/9554df5/ : SUCCESS in 51m 20s\n- patrole-member http://logs.openstack.org/37/670637/2/gate/patrole-member/63d3bba/ : SUCCESS in 1h 09m 39s","accounts_in_message":[],"_revision_number":2},{"id":"140737df4384aaa57f87f0f67f3c25765a5aaefa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 21:39:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f15d30d75df8cf4e6019719018f6982370c18be","revisions":{"498828cc05d3c48c1cad5d5806a40b8ef524a459":{"kind":"REWORK","_number":1,"created":"2019-07-12 20:08:04.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/37/670637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/37/670637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/37/670637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/37/670637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/37/670637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/37/670637/1"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:55:31.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-12 20:08:03.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ib3e0b4220d9a8a076ba2c26ac68852aa3953f134\nStory: #2005924\nTask: #34240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/498828cc05d3c48c1cad5d5806a40b8ef524a459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/498828cc05d3c48c1cad5d5806a40b8ef524a459"}]},"branch":"refs/heads/master"},"6f15d30d75df8cf4e6019719018f6982370c18be":{"kind":"REWORK","_number":2,"created":"2019-07-15 17:19:51.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/37/670637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/37/670637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/37/670637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/37/670637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/37/670637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/37/670637/2"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:55:31.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-15 17:19:43.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ib3e0b4220d9a8a076ba2c26ac68852aa3953f134\nStory: #2005924\nTask: #34240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6f15d30d75df8cf4e6019719018f6982370c18be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/6f15d30d75df8cf4e6019719018f6982370c18be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
