)]}'
{"id":"openstack%2Fpatrole~670702","triplet_id":"openstack%2Fpatrole~master~I71c211f91c320299c873652f46acd5cb93afb825","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I71c211f91c320299c873652f46acd5cb93afb825","subject":"Add releasenote to tag the end of support for Pike","status":"MERGED","created":"2019-07-13 16:10:20.000000000","updated":"2019-07-23 17:04:08.000000000","submitted":"2019-07-23 17:04:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670702-1563901448196-8239778b","meta_rev_id":"42799a40fed02d7b7d3fbe3ec1f29b261807a974","_number":670702,"virtual_id_number":670702,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2019-07-23 17:04:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2019-07-18 03:07:44.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-22 08:43:45.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-22 08:43:45.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 03:07:44.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-07-23 17:04:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24030005ccc7029a8924e59f8fbde5a2be2880d5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-13 16:10:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c868eb4706e4db5821bcd6c5cc1de1a2f64ad84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 16:13:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/02/670702/1/check/openstack-tox-pep8/a9d9e4f/ : SUCCESS in 2m 34s\n- openstack-tox-docs http://logs.openstack.org/02/670702/1/check/openstack-tox-docs/f356b6e/html/ : SUCCESS in 2m 14s\n- build-openstack-releasenotes http://logs.openstack.org/02/670702/1/check/build-openstack-releasenotes/a8b7794/ : FAILURE in 2m 35s","accounts_in_message":[],"_revision_number":1},{"id":"77f6daa94450614e12d4e66ced961f525ff85098","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-07-15 18:43:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fd46f694d5774e64c86ae0cfc79606b1f60807f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 19:03:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/02/670702/2/check/openstack-tox-pep8/6560474/ : SUCCESS in 2m 57s\n- openstack-tox-docs http://logs.openstack.org/02/670702/2/check/openstack-tox-docs/73a04ac/html/ : SUCCESS in 3m 06s\n- build-openstack-releasenotes http://logs.openstack.org/02/670702/2/check/build-openstack-releasenotes/a695ce6/ : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"61db4a50326748b241bc3436aeb5e10bcf8f6ae7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-17 13:00:01.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"bde615e8a1e738c0016814ac681a94d015a4cf5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 13:09:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/02/670702/3/check/openstack-tox-pep8/8ec2a2c/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/02/670702/3/check/openstack-tox-docs/cabaff1/html/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/02/670702/3/check/build-openstack-releasenotes/0cfa835/html/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d3dcb2dbec3829cfd809bf543167f629ec2f5ce3","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-07-18 03:07:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1287fe7074eb86d369f5f32282b0b4abcc8c535c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-22 08:43:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nlet\u0027s get this in to release patrole.","accounts_in_message":[],"_revision_number":3},{"id":"17164ec88ea4941de23fa644b2e85b50d7cddc0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 16:53:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"57c56b99fd6757714a0f16b17e30dce74e3ce958","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:04:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/02/670702/3/gate/openstack-tox-pep8/117aee5/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/02/670702/3/gate/openstack-tox-docs/fc1dc8e/html/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/02/670702/3/gate/build-openstack-releasenotes/7afa22b/html/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"42799a40fed02d7b7d3fbe3ec1f29b261807a974","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:04:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bad5c037bde3484f540455bd4a094df3178f0a5d","revisions":{"5978d0be835633137264cf264f78684473972131":{"kind":"REWORK","_number":1,"created":"2019-07-13 16:10:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/670702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/02/670702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/02/670702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/02/670702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/02/670702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/02/670702/1"}}},"commit":{"parents":[{"commit":"5b5f1b1dc72344ba591c554d7c195245e59c17c8","subject":"Merge \"Fix some inaccurate URL links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5b5f1b1dc72344ba591c554d7c195245e59c17c8"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-07-13 16:09:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-07-13 16:09:33.000000000","tz":0},"subject":"Add releasenote to tag the end of support for Pike","message":"Add releasenote to tag the end of support for Pike\n\nThis relesenote also mark end of support for Pike release\nin Patrole as Pike is in EM state[1].\n\n[1] https://releases.openstack.org/\n\nChange-Id: I71c211f91c320299c873652f46acd5cb93afb825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5978d0be835633137264cf264f78684473972131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5978d0be835633137264cf264f78684473972131"}]},"branch":"refs/heads/master"},"84b94ba934ee1cbe3e055398b7ae39b6840735bd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-15 18:43:37.000000000","uploader":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"ref":"refs/changes/02/670702/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/02/670702/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/02/670702/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/02/670702/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/02/670702/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/02/670702/2"}}},"commit":{"parents":[{"commit":"5bb8f59557c6fb403d997d1d131cc508bf02d909","subject":"Revert \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5bb8f59557c6fb403d997d1d131cc508bf02d909"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-07-13 16:09:33.000000000","tz":0},"committer":{"name":"Felipe Monteiro","email":"felipe.monteiro@att.com","date":"2019-07-15 18:43:37.000000000","tz":0},"subject":"Add releasenote to tag the end of support for Pike","message":"Add releasenote to tag the end of support for Pike\n\nThis relesenote also mark end of support for Pike release\nin Patrole as Pike is in EM state[1].\n\n[1] https://releases.openstack.org/\n\nChange-Id: I71c211f91c320299c873652f46acd5cb93afb825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/84b94ba934ee1cbe3e055398b7ae39b6840735bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/84b94ba934ee1cbe3e055398b7ae39b6840735bd"}]},"branch":"refs/heads/master"},"bad5c037bde3484f540455bd4a094df3178f0a5d":{"kind":"REWORK","_number":3,"created":"2019-07-17 13:00:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/670702/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/02/670702/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/02/670702/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/02/670702/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/02/670702/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/02/670702/3"}}},"commit":{"parents":[{"commit":"5bb8f59557c6fb403d997d1d131cc508bf02d909","subject":"Revert \"Update master for stable/stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5bb8f59557c6fb403d997d1d131cc508bf02d909"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-07-13 16:09:33.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-07-17 12:59:58.000000000","tz":0},"subject":"Add releasenote to tag the end of support for Pike","message":"Add releasenote to tag the end of support for Pike\n\nThis relesenote also mark end of support for Pike release\nin Patrole as Pike is in EM state[1].\n\n[1] https://releases.openstack.org/\n\nChange-Id: I71c211f91c320299c873652f46acd5cb93afb825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/bad5c037bde3484f540455bd4a094df3178f0a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/bad5c037bde3484f540455bd4a094df3178f0a5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
