)]}'
{"id":"openstack%2Fpatrole~671872","triplet_id":"openstack%2Fpatrole~master~I3ed450ab312af81c057659f47f1ba0241d6aad23","project":"openstack/patrole","branch":"master","hashtags":[],"change_id":"I3ed450ab312af81c057659f47f1ba0241d6aad23","subject":"Make patrole-member-stein non-voting","status":"ABANDONED","created":"2019-07-19 22:53:57.000000000","updated":"2019-09-19 17:53:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b319030b7c750b4b21b78fca233b47bfed6904ce","_number":671872,"virtual_id_number":671872,"owner":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-07-20 00:22:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":-1,"date":"2019-07-22 23:47:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":-1,"date":"2019-09-10 13:03:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-06 02:22:29.000000000","_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-20 00:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-22 23:47:13.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-08-06 02:22:29.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2019-09-10 13:03:14.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"9e4ca577ed78a3e29f8722feff9eadaa587fc2d6","author":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"date":"2019-07-19 22:53:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11a3505cd376b73e0eefe03503effd470653049e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 00:22:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/72/671872/1/check/openstack-tox-cover/b7e1d15/cover/ : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/671872/1/check/openstack-tox-lower-constraints/69793d2/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/72/671872/1/check/openstack-tox-py36/e2de131/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/72/671872/1/check/openstack-tox-py37/81b7f3c/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/72/671872/1/check/openstack-tox-pep8/3e9fef1/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/72/671872/1/check/openstack-tox-py27/2658cb3/ : SUCCESS in 2m 47s\n- openstack-tox-docs http://logs.openstack.org/72/671872/1/check/openstack-tox-docs/b373ed3/html/ : SUCCESS in 3m 19s\n- patrole-admin http://logs.openstack.org/72/671872/1/check/patrole-admin/ac4aff3/ : SUCCESS in 58m 55s\n- patrole-member http://logs.openstack.org/72/671872/1/check/patrole-member/e650474/ : SUCCESS in 1h 03m 10s\n- patrole-reader http://logs.openstack.org/72/671872/1/check/patrole-reader/de93c82/ : SUCCESS in 1h 03m 11s (non-voting)\n- patrole-member-stein http://logs.openstack.org/72/671872/1/check/patrole-member-stein/9bf9eed/ : FAILURE in 1h 11m 00s (non-voting)\n- patrole-member-rocky http://logs.openstack.org/72/671872/1/check/patrole-member-rocky/7839382/ : SUCCESS in 1h 11m 41s\n- patrole-member-queens http://logs.openstack.org/72/671872/1/check/patrole-member-queens/908cde9/ : SUCCESS in 1h 13m 16s\n- patrole-member-pike http://logs.openstack.org/72/671872/1/check/patrole-member-pike/c96eb7d/ : SUCCESS in 1h 22m 20s\n- patrole-multinode-admin http://logs.openstack.org/72/671872/1/check/patrole-multinode-admin/cb23cb2/ : FAILURE in 35m 44s (non-voting)\n- patrole-multinode-member http://logs.openstack.org/72/671872/1/check/patrole-multinode-member/8b5b30e/ : FAILURE in 32m 43s (non-voting)\n- patrole-extension-admin http://logs.openstack.org/72/671872/1/check/patrole-extension-admin/c44c2f1/ : SUCCESS in 36m 42s (non-voting)\n- patrole-extension-member http://logs.openstack.org/72/671872/1/check/patrole-extension-member/4d627db/ : FAILURE in 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ef203bf9ad98f354bdcb16cc52373aa34737039","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-07-22 23:47:13.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for taking the initiative to fix these failures. But the right way to avoid failing this job is to add a feature flag to tests like:\n\n* test_update_network_provider_physical_network\n* test_update_network_provider_segmentation_id\n\nExample: https://github.com/openstack/patrole/blob/c8246e66e45a710e3c4ef427eeef845b71dd343f/devstack/plugin.sh#L59\n\nThat skips them if stein is enabled.\n\nBut I suspect the best way is to simply fix the above two tests (or as a last resort remove them if they are no longer applicable). Skipping the entire job to get around the fact that these 2 tests are the problematic ones is not the right approach. We should keep this as voting.","accounts_in_message":[],"_revision_number":1},{"id":"70c50656aee850dd97c9b4bdb920e115e33cfd06","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-10 13:03:14.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Thanks for taking the initiative to fix these failures. But the\n \u003e right way to avoid failing this job is to add a feature flag to\n \u003e tests like:\n \u003e \n \u003e * test_update_network_provider_physical_network\n \u003e * test_update_network_provider_segmentation_id\n \u003e \n \u003e Example: https://github.com/openstack/patrole/blob/c8246e66e45a710e3c4ef427eeef845b71dd343f/devstack/plugin.sh#L59\n \u003e \n \u003e That skips them if stein is enabled.\n \u003e \n \u003e But I suspect the best way is to simply fix the above two tests (or\n \u003e as a last resort remove them if they are no longer applicable).\n \u003e Skipping the entire job to get around the fact that these 2 tests\n \u003e are the problematic ones is not the right approach. We should keep\n \u003e this as voting.\n\n+1 let\u0027s fix the test via feature flag instead of job.","accounts_in_message":[],"_revision_number":1},{"id":"b319030b7c750b4b21b78fca233b47bfed6904ce","author":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"date":"2019-09-19 17:53:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ec04b8361a7ddc610a8adcc18e836cde0c86f05","revisions":{"1ec04b8361a7ddc610a8adcc18e836cde0c86f05":{"kind":"REWORK","_number":1,"created":"2019-07-19 22:53:57.000000000","uploader":{"_account_id":23140,"name":"Jess Egler","email":"jess.egler@gmail.com","username":"jessegler"},"ref":"refs/changes/72/671872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/72/671872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/72/671872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/72/671872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/72/671872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/72/671872/1"}}},"commit":{"parents":[{"commit":"c8246e66e45a710e3c4ef427eeef845b71dd343f","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/c8246e66e45a710e3c4ef427eeef845b71dd343f"}]}],"author":{"name":"jessegler","email":"jess.egler@gmail.com","date":"2019-07-19 22:51:24.000000000","tz":-300},"committer":{"name":"jessegler","email":"jess.egler@gmail.com","date":"2019-07-19 22:51:24.000000000","tz":-300},"subject":"Make patrole-member-stein non-voting","message":"Make patrole-member-stein non-voting\n\nThe patrole-member-stein job is failing intermittently, so this patchset\nmakes it non-voting until it can be removed or fixed.\n\nChange-Id: I3ed450ab312af81c057659f47f1ba0241d6aad23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1ec04b8361a7ddc610a8adcc18e836cde0c86f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1ec04b8361a7ddc610a8adcc18e836cde0c86f05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
