)]}'
{"id":"openstack%2Fpatrole~676979","triplet_id":"openstack%2Fpatrole~master~I88d0624bc279b004ebe5f721cdef5adf12bd0ed9","project":"openstack/patrole","branch":"master","topic":"neutron_extension_tests_not_working_with_config","attention_set":{},"removed_from_attention_set":{"17896":{"account":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"last_update":"2023-03-11 02:13:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I88d0624bc279b004ebe5f721cdef5adf12bd0ed9","subject":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests","status":"ABANDONED","created":"2019-08-16 16:02:59.000000000","updated":"2023-03-11 02:13:23.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f93ed99ff4d8b4eb03431ab0d59800905107bead","_number":676979,"virtual_id_number":676979,"owner":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"date":"2019-10-07 23:52:29.000000000","_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-09-05 01:23:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":24817,"name":"Avishek Dutta","email":"ad620p@att.com","username":"ad620p"},{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":-1,"date":"2019-09-20 20:00:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24817,"name":"Avishek Dutta","email":"ad620p@att.com","username":"ad620p"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"date":"2019-10-02 21:50:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"date":"2020-02-11 13:12:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":24817,"name":"Avishek Dutta","email":"ad620p@att.com","username":"ad620p"},{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},{"_account_id":24817,"name":"Avishek Dutta","email":"ad620p@att.com","username":"ad620p"},{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"reviewer":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"state":"REVIEWER"},{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"reviewer":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"state":"REVIEWER"},{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"reviewer":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"state":"REVIEWER"},{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2019-08-16 16:04:14.000000000","updated_by":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"reviewer":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"state":"REVIEWER"},{"updated":"2019-08-21 16:01:04.000000000","updated_by":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"reviewer":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"state":"REVIEWER"},{"updated":"2019-08-21 16:01:09.000000000","updated_by":{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},"reviewer":{"_account_id":30745,"name":"Kelly Carter","email":"KC755D@att.com","username":"kc755d"},"state":"REVIEWER"},{"updated":"2019-08-30 13:47:40.000000000","updated_by":{"_account_id":24817,"name":"Avishek Dutta","email":"ad620p@att.com","username":"ad620p"},"reviewer":{"_account_id":24817,"name":"Avishek Dutta","email":"ad620p@att.com","username":"ad620p"},"state":"REVIEWER"},{"updated":"2019-09-05 01:23:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-20 20:00:15.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-10-02 21:50:44.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-02-11 13:12:11.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"}],"messages":[{"id":"b1f9d06f3c1e85cfe54e1f34bd8214db3dfd72c8","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-08-16 16:02:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5aa88fe352ba15c232f0d2b3afa78f370e6a203","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 17:24:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3092d54d68a4024ac74cb9c79a054c9 : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f42c0af4b8b44f3a5473ace17ff0a43 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1399383e13fe46f1a4d704cfc60356bf : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdfafa40443f4634b00ef3a7ce0f207f : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20879fe38c954bd3b43db91bbb81b9b7 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c1c5512c9af447d830aad7b18b6f210 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f38ad7cc3424fa9b4d2ab1404922516 : SUCCESS in 3m 55s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/e2b318a5b414476bab1b14aea2e69936 : SUCCESS in 49m 01s\n- patrole-member https://zuul.opendev.org/t/openstack/build/60451ad009c34fc69cb6dd17262833d9 : SUCCESS in 1h 09m 21s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c6489eaed5474f108c139681220e5e60 : SUCCESS in 1h 12m 52s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/e71314cf84364fae964e68e3c9ff2813 : SUCCESS in 1h 15m 54s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/819e03426e77408996892787f49bcf03 : SUCCESS in 1h 20m 28s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/830d65aedc4345cfa00f9fbffc864bdd : SUCCESS in 1h 13m 59s\n- patrole-member-pike https://zuul.opendev.org/t/openstack/build/5dd974c3b9074fe2bc4f16160c03da16 : SUCCESS in 1h 10m 09s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/36157cb4b7354038842003a37ea3c091 : FAILURE in 1h 17m 50s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/94cd3207f1e145d885748d58a875af7f : SUCCESS in 1h 06m 09s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/938104cb182446b7990d61c79005a60d : SUCCESS in 36m 34s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/65175ee306934995a6f34e52269728ae : FAILURE in 39m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3391372f308b4861862d7a5176e4e1431380c6f9","author":{"_account_id":24817,"name":"Avishek Dutta","email":"ad620p@att.com","username":"ad620p"},"date":"2019-08-20 06:34:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9289a63c0bab7ba6ee8872780fd585cef4836296","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-08-21 16:20:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bdfd94c8c472502e24e07b05099b3a039d78267d","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-08-23 21:27:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4e13093474e6eddc5271868bb8aa4afd883dc3d3","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-08-23 21:29:33.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nHey Felipe, thanks for the review. Great point on copying every attribute. Frankly I am not sure why the creds are lost. I tried to trace it through and couldn\u0027t find the exact reason.","accounts_in_message":[],"_revision_number":2},{"id":"9c07ac5ad1c234865a9a739f5c668f7859dbd8ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 22:53:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b468b4a07bb64c849bf8967a44026d3c : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28f86b43fd9844b09c28115ac05cb411 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73d544b31f5d4fe9bdc1e7aa0ce91fbe : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fee1382977bd4fa7a12fb6fa58ccb6e0 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/549e5c6d6e514f26b4178a8f3fd5e196 : FAILURE in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6250eff3d0842f7a353f8f05bc230a5 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da5435e04e34989b7456e32db13f10a : FAILURE in 3m 49s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/10ffa815162a4b829999ed33570c1c8a : SUCCESS in 1h 11m 36s\n- patrole-member https://zuul.opendev.org/t/openstack/build/2d594f80df1c41ea9a559940820456d4 : SUCCESS in 1h 13m 33s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/f4363956d89949ebb66550f3420afcdc : SUCCESS in 1h 13m 53s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/6487368dc7794041bb7be560d0503dfe : SUCCESS in 1h 04m 34s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/bc6bc7c8a587418eaa177ce817293633 : SUCCESS in 1h 12m 57s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/75d474a5b7c04c40965fdd50735438c5 : SUCCESS in 1h 01m 42s\n- patrole-member-pike https://zuul.opendev.org/t/openstack/build/b8d98dac1425461c8430b8e3f0da652e : SUCCESS in 1h 15m 36s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/8692e2369b1c4836ac04acaba7f4f6e4 : FAILURE in 1h 20m 15s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/e67ebef93e9549ffade8f114f228b2ec : SUCCESS in 1h 02m 10s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/05895d2bd2d04fbabaf46aa9c2584feb : SUCCESS in 37m 36s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/179b26ee2f1b49f4a4a08ccec69eb9aa : FAILURE in 39m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd88030ac900c6cbc1e2848096895b6bece53814","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-08-26 19:42:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f91770ebb8f2921c0348a49ba1f87ed6f35e7a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 21:10:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0602e96925e947ff9ee4e0caee248303 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bf1eed6d9a248ed9c4a7b68be9f406a : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41bb5dc437354e33a83293db492a729d : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b33a702e453f49cca0d903cb65886ee3 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db21c813daa4aa68a43960560ec301f : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea0a31ea28774b82877d72a81cb042f7 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3449f3685f54915a0cf87c7f512ea23 : FAILURE in 3m 59s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/79f86862261b493cb1c4bb39696a3d00 : SUCCESS in 59m 03s\n- patrole-member https://zuul.opendev.org/t/openstack/build/0f6236d842ce43989ef77a2dfb8b3c80 : SUCCESS in 1h 11m 16s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/9a1ef07b2cce4339a6ac4e8b3f101970 : SUCCESS in 1h 08m 09s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/2ab7abcbb86a46d782742ad625303ef5 : SUCCESS in 1h 03m 58s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/4940f5dd83be4af4a8e24fa829d1b40d : SUCCESS in 1h 27m 17s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/77948cf4cf5d46ffa93bf6eec98b5f77 : SUCCESS in 57m 12s\n- patrole-member-pike https://zuul.opendev.org/t/openstack/build/67e026a7599040ceb127837d0d221c99 : SUCCESS in 1h 07m 10s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/443ef40ccfbf4f71a04211e41c20083c : FAILURE in 1h 12m 19s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/cac57ccb12994082bcec510e2fca1e61 : SUCCESS in 1h 06m 43s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/ffdba525dcd44357a0e0c83da4469618 : SUCCESS in 37m 55s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/9950be5a941d44e8a8f9f52c0a2765a4 : FAILURE in 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0c7adc60c27e881dc3c75ff83df92d9e40d654c","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-08-27 14:46:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"efa2bd47ed4ad1b60d31ee1f1abfc5d3a7484bf5","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-08-27 17:57:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"18aecf1315d4faca5880e642da2bd7561c070c3f","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-08-27 17:57:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e4fb11015760190360c79be8cc6ea504cbb87713","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-08-27 18:01:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"859661fbe290d30660a9567a06a24d3548199cd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 19:49:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6b6e46ae0fb49908c2f4ee6cf3c4303 : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/106fc13b7e0b46fc849e35f0178d27c4 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8554552f0a2f40e2a862e464b598bc74 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f715c7ca851448519181a6e0371804e6 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4e64b934810436fa6fa4bc8914e5dab : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56eac920eeff4760a09b6d34553148b7 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc1c22022cd7497aa7bbddd9b9bf110a : FAILURE in 4m 26s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/e4cb0058e094467d955d022d586a9b70 : SUCCESS in 1h 01m 06s\n- patrole-member https://zuul.opendev.org/t/openstack/build/1d978269e28448969f0a49940601c8dd : SUCCESS in 1h 05m 31s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/34ad435b179542bea6218b66e1def511 : SUCCESS in 1h 13m 01s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/3dac1ff0d5f54ba892d3c611d85aaa80 : SUCCESS in 1h 24m 03s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/c832bfaa99294f468e9d16528e6b5715 : SUCCESS in 1h 22m 13s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/77d93b47dfe34eed867f416e5ba25df2 : SUCCESS in 1h 06m 25s\n- patrole-member-pike https://zuul.opendev.org/t/openstack/build/fe9f440d76214321b25d21959dae3ffd : SUCCESS in 1h 06m 51s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c1a91ac51f1648dcae798fa9ac7ec445 : FAILURE in 1h 17m 40s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/98a7cdd26d0a47ca9cb61e34cd5d2e0a : SUCCESS in 1h 10m 24s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/b9a24fdfea4044b881db921741b0f6e7 : SUCCESS in 41m 23s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/8d8794be961e4bd494d2634f3127e266 : FAILURE in 39m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"57d04d4e7125b81cede36c4196ff179895d18a40","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-08-29 13:14:56.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a8d4c1ccee1393e1608b1ed78410976e7de3c5cd","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-08-29 14:07:37.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks, LGTM.","accounts_in_message":[],"_revision_number":5},{"id":"ebbe1b0d536c23d993b45612a7e6187ed60adf45","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-08-29 14:42:34.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"ffeb95f558c4f0d87d01f3b6f5094c51a65482b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:23:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10cfb943dab34164b984855e3739b168 : SUCCESS in 3m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/457f7c70e7e642c087430e578293da52 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23c41bb1741c40ceb13763b39fee639a : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf116c347bf14830b37902bd2ca0d064 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b20b332192b417491ebfb9f71d5b537 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f384bd38bd64e65ae03ec0ff0554783 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b63b426be84052a5bf6f3f810511ee : SUCCESS in 3m 34s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/5d224d1b32f247e39fa626d87df9517e : FAILURE in 1h 11m 00s\n- patrole-member https://zuul.opendev.org/t/openstack/build/8bded61d3f68423c8b53629eb90a5556 : SUCCESS in 1h 16m 50s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/1785999c039841298e8fac64f4b8c0ba : SUCCESS in 1h 18m 46s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/db8e5926d4c34c95b4909d9ec13f6d4c : FAILURE in 1h 05m 24s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/044c25b204e94ee5ba7be7cfb62a8410 : SUCCESS in 1h 13m 42s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/b1fdcbd7fea942ec9a18a59111d6c005 : FAILURE in 1h 21m 35s\n- patrole-member-pike https://zuul.opendev.org/t/openstack/build/204c20e5fe0b471797c877297938ee1b : SUCCESS in 1h 28m 01s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/b7df1a0ba1b34bd6bf434273329e5495 : FAILURE in 1h 12m 25s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/8af506fd847f4612864ecf6227e377b5 : SUCCESS in 1h 15m 13s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/5b25896f7f334c45a58d527696625655 : SUCCESS in 40m 26s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/0719b50965184303a63d67b61229d667 : FAILURE in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"48de53591c793fee3842f960bb5020264d6274dc","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-08-30 11:32:03.000000000","message":"Patch Set 6:\n\nRick this is failing due to new response validation \nhttps://docs.openstack.org/api-ref/block-storage/v3/?expanded\u003dshow-host-details-for-a-project-detail#show-host-details-for-a-project\nthe API states that the name should be xxx@yyy\nbut the test has no @","accounts_in_message":[],"_revision_number":6},{"id":"11ae4214d0296ed27199ca331f4a87c764969236","author":{"_account_id":24817,"name":"Avishek Dutta","email":"ad620p@att.com","username":"ad620p"},"date":"2019-08-30 13:47:40.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7bb93654374458d145fd3632f437b93a75d0e4f2","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-09-04 18:47:50.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c85160d8ad80140f73c817f45c1c332f5c8c9c30","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-09-04 18:52:19.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"113f269468bc1bb6f7e9f065315f0e5b6710f652","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 20:45:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fef65dbafd8f48a2ad7bd37972cf0995 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72ad837f80644a12b82802d82890b2f2 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/007abfd79cb14c6691c4b179affa7488 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d8aa8ecabf0419886aff582e7f29697 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ee0d33dd3884041bb1f3b3a984e5024 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e96d828f06a24f6a83d5af4a2ea230d5 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0de5033e73d4a7c90f6af60c26e38cd : SUCCESS in 7m 48s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/bef0c0208b364cf99c2eb94a5a8d3a7b : FAILURE in 1h 28m 32s\n- patrole-member https://zuul.opendev.org/t/openstack/build/07e7bc09de1346a2a09ea28f0337083d : SUCCESS in 1h 07m 15s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/806faf4769514b1eb5eaada35334b008 : SUCCESS in 1h 49m 54s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/930e5d10e7924d778d60c275f8b5abb4 : FAILURE in 1h 07m 47s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/f57dfcd1dbbe4255a70225a97ba822bc : SUCCESS in 1h 08m 38s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/b49ab43ba93042ae9e352ec7c6d2cef6 : SUCCESS in 1h 07m 06s\n- patrole-member-pike https://zuul.opendev.org/t/openstack/build/12a37c97453d45f597cf022fb9570e52 : FAILURE in 1h 09m 07s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/257c2a16257840f690723dbd1d178595 : FAILURE in 1h 24m 08s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/ef12356cbe934771aef5c1efd2e1543b : SUCCESS in 1h 08m 02s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/0917c756155d4672bed2833e4b151fe3 : SUCCESS in 41m 15s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/d9c76a431d204cc497be54043a2b0ebd : FAILURE in 38m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"71bffd906a7e7612ece8284441b57e44e571e17d","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-09-04 23:43:24.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"96cec81aa59dd384db5072b9bdf5a4c26490e7b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 01:23:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9ced250de6543ef964987877ef99d48 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65d20912d2e04ffbbccac8bd9c365352 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c413b29805a94348b1be7b120bf55332 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b473bba7776c470a8aad88f80910b487 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c2d5046d4c44a7ae2b5991864afaf3 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3a37db68b3a4286a74f40d50e3f13c3 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536ef31c6718411f8f05d54fa9438b10 : SUCCESS in 5m 15s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/187c75cb92d64fba89cc86ffe205ee7d : SUCCESS in 53m 34s\n- patrole-member https://zuul.opendev.org/t/openstack/build/3f4984a83fe84a92bf95b9db29251bd5 : SUCCESS in 1h 21m 23s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/2f9d2a12dacd49e4a05b9e2ab8ab17da : SUCCESS in 1h 14m 45s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/7e46d8ea989d44f28aefe90a0cd06d9b : SUCCESS in 1h 16m 05s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/0be0c70e6af64dddb58a3c050f75d7b3 : SUCCESS in 1h 01m 23s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/6cc352b4a1d14231ab9bb107df4c8c34 : SUCCESS in 1h 33m 11s\n- patrole-member-pike https://zuul.opendev.org/t/openstack/build/136c01fb2abd47a188886a0727ad6ee4 : SUCCESS in 1h 23m 46s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/862decebd03e411183cdd472e06a8a62 : FAILURE in 1h 10m 47s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/5f16e47f211e42079c584b116c010c1d : SUCCESS in 1h 10m 03s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/2493a06b6b244f12943b865c5d4bdc7b : SUCCESS in 35m 10s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/b24b843f0d5245e896c2dcbfbe6f47e0 : FAILURE in 40m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0d835d67501deb6d48a81151235c040a73a56c45","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-09-05 05:24:46.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"bce19d1c0670bf388ed6afcb9106c0741643e35d","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-09-05 10:57:14.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"a0567c49663ff85b18683e56a0716abc0cea85ca","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-20 20:00:15.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nlet\u0027s find the real cause of this issue. This seems interesting to me and might find out hidden bug in Tempest. let\u0027s not workaround that.","accounts_in_message":[],"_revision_number":7},{"id":"e48e743c1d16989bee94245b0388fe18f0771287","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-10-02 20:52:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"54db5742a91c9d42fd56b2d13b586bcfa08b3dd8","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-10-02 21:50:44.000000000","message":"Patch Set 7: -Code-Review\n\nAgree with gmann that the best approach is to fix this issue Tempest side rather than implement a workaround which can obscure the true root cause.","accounts_in_message":[],"_revision_number":7},{"id":"c6e2ce33157c20c531629fd8a0a77d286fa3c765","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-10-03 21:41:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7793a6047ccffa8fec589a2cf411a09d6d888b30","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-10-07 23:52:29.000000000","message":"Patch Set 7:\n\n\u0027user_id\u0027 and \u0027tenant_id\u0027 are not part of Credentials self._initial (https://github.com/openstack/tempest/blob/master/tempest/lib/auth.py#L667) and therefore are not restored when self.clear_auth is called:\n\nThe reason why they are not part of self._initial is because with pre-provisioned creds, these fields are not in the accounts.yaml that we use. They can be added to the accounts.yaml manually, but that is overkill.\n\nWhen self.clear_auth is called:\nhttps://github.com/openstack/tempest/blob/master/tempest/lib/auth.py#L251\nthe credentials are reset, and only those fields that belong to the Credentials self._initial are restored in the \u0027_apply_credentials\u0027 method: https://github.com/openstack/tempest/blob/master/tempest/lib/auth.py#L678\n\nThe Credentials reset method could be re-written in https://github.com/openstack/tempest/blob/master/tempest/lib/auth.py to force self._initial to include all keys that were defined in the Credentials object:\n\n    def reset(self):\n        # First delete all known attributes\n        for key in self.ATTRIBUTES:\n            if getattr(self, key) is not None:\n                # add all known attributes to self._initial so that no cred attributes are \n                # missing when _apply_credentials is called\n                self._initial.update({key: getattr(self, key)})\n                delattr(self, key)\n        # Then re-apply initial setup\n        self._apply_credentials(self._initial)\n\nAny other suggestion on how to fix in Tempest would be helpful. I do not think what I am proposing above will be accepted in Tempest and especially into Tempest lib. Until a fix is made in Tempest, can this please be merged as a workaround then later reverted when someone can fix the \u0027bug\u0027 in Tempest? I am not completely sure this is even a bug in Tempest especially given this: https://github.com/openstack/tempest/blob/master/tempest/lib/auth.py#L664","accounts_in_message":[],"_revision_number":7},{"id":"635cf0771525d256c21f74a720ecd658b05f2dfd","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-02-11 13:12:11.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"006ea63ce5e269a903fecade15df31715e4770dc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-11 02:12:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f93ed99ff4d8b4eb03431ab0d59800905107bead","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-11 02:13:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e88d55248ad9aaa9876c906e533e0d456eb54520","revisions":{"5cd43821446fb4f4d72d88e67925db4556f90d24":{"kind":"REWORK","_number":1,"created":"2019-08-16 16:02:59.000000000","uploader":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"ref":"refs/changes/79/676979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/79/676979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/79/676979/1"}}},"commit":{"parents":[{"commit":"17fc33340dd6c449ce504e70b357e914bcd0a919","subject":"Fix a broken url link in README.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/17fc33340dd6c449ce504e70b357e914bcd0a919"}]}],"author":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-16 15:42:54.000000000","tz":-240},"committer":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-16 15:42:54.000000000","tz":-240},"subject":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests","message":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests\n\nWhen running Neutron extension tests and Tempest configured with\n\u0027use_dynamic_credentials \u003d false\u0027 the user_id and tenant_id are unset\nand results in the tests not being executed: \"Could not find user: None.\"\n\nBoth user_id and tenant_id are needed in rbac_utils [0]\n\n[0] https://github.com/openstack/patrole/blob/master/patrole_tempest_plugin/rbac_utils.py#L163\n\nChange-Id: I88d0624bc279b004ebe5f721cdef5adf12bd0ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5cd43821446fb4f4d72d88e67925db4556f90d24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5cd43821446fb4f4d72d88e67925db4556f90d24"}]},"branch":"refs/heads/master"},"2e956ead0096fc44c38bdfd7db2cf456447cca4c":{"kind":"REWORK","_number":2,"created":"2019-08-23 21:27:52.000000000","uploader":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"ref":"refs/changes/79/676979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/79/676979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/79/676979/2"}}},"commit":{"parents":[{"commit":"78af751e7db7f138622512b5a1914e5aae4185e5","subject":"Merge \"Add release notes page for version 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/78af751e7db7f138622512b5a1914e5aae4185e5"}]}],"author":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-16 15:42:54.000000000","tz":-240},"committer":{"name":"Rick Bartra","email":"rb560u@att.com","date":"2019-08-23 21:27:12.000000000","tz":-240},"subject":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests","message":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests\n\nWhen running Neutron extension tests and Tempest configured with\n\u0027use_dynamic_credentials \u003d false\u0027 the user_id and tenant_id are unset\nand results in the tests not being executed: \"Could not find user: None.\"\n\nBoth user_id and tenant_id are needed in rbac_utils [0]\n\n[0] https://github.com/openstack/patrole/blob/master/patrole_tempest_plugin/rbac_utils.py#L163\n\nChange-Id: I88d0624bc279b004ebe5f721cdef5adf12bd0ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2e956ead0096fc44c38bdfd7db2cf456447cca4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2e956ead0096fc44c38bdfd7db2cf456447cca4c"}]},"branch":"refs/heads/master"},"2b21d5203aec7ab35c90bbdd8aedd995e7a83d7e":{"kind":"REWORK","_number":3,"created":"2019-08-26 19:42:33.000000000","uploader":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"ref":"refs/changes/79/676979/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/79/676979/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/79/676979/3"}}},"commit":{"parents":[{"commit":"78af751e7db7f138622512b5a1914e5aae4185e5","subject":"Merge \"Add release notes page for version 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/78af751e7db7f138622512b5a1914e5aae4185e5"}]}],"author":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-16 15:42:54.000000000","tz":-240},"committer":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-26 19:38:36.000000000","tz":-240},"subject":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests","message":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests\n\nWhen running Neutron extension tests and Tempest configured with\n\u0027use_dynamic_credentials \u003d false\u0027 the user_id and tenant_id are unset\nand results in the tests not being executed: \"Could not find user: None.\"\n\nBoth user_id and tenant_id are needed in rbac_utils [0]\n\n[0] https://github.com/openstack/patrole/blob/master/patrole_tempest_plugin/rbac_utils.py#L163\n\nChange-Id: I88d0624bc279b004ebe5f721cdef5adf12bd0ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2b21d5203aec7ab35c90bbdd8aedd995e7a83d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2b21d5203aec7ab35c90bbdd8aedd995e7a83d7e"}]},"branch":"refs/heads/master"},"4c5040484387b4fd348c20b518d898387a6febde":{"kind":"REWORK","_number":4,"created":"2019-08-27 17:57:31.000000000","uploader":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"ref":"refs/changes/79/676979/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/79/676979/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/79/676979/4"}}},"commit":{"parents":[{"commit":"78af751e7db7f138622512b5a1914e5aae4185e5","subject":"Merge \"Add release notes page for version 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/78af751e7db7f138622512b5a1914e5aae4185e5"}]}],"author":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-16 15:42:54.000000000","tz":-240},"committer":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-27 17:55:32.000000000","tz":-240},"subject":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests","message":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests\n\nWhen running Neutron extension tests and Tempest configured with\n\u0027use_dynamic_credentials \u003d false\u0027 the user_id and tenant_id are unset\nand results in the tests not being executed: \"Could not find user: None.\"\n\nBoth user_id and tenant_id are needed in rbac_utils [0]\n\n[0] https://github.com/openstack/patrole/blob/master/patrole_tempest_plugin/rbac_utils.py#L163\n\nChange-Id: I88d0624bc279b004ebe5f721cdef5adf12bd0ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4c5040484387b4fd348c20b518d898387a6febde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4c5040484387b4fd348c20b518d898387a6febde"}]},"branch":"refs/heads/master"},"2bd8e6f1ee4129ae2a38827993bbbad4051158b4":{"kind":"REWORK","_number":5,"created":"2019-08-27 18:01:21.000000000","uploader":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"ref":"refs/changes/79/676979/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/79/676979/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/79/676979/5"}}},"commit":{"parents":[{"commit":"78af751e7db7f138622512b5a1914e5aae4185e5","subject":"Merge \"Add release notes page for version 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/78af751e7db7f138622512b5a1914e5aae4185e5"}]}],"author":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-16 15:42:54.000000000","tz":-240},"committer":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-27 18:00:34.000000000","tz":-240},"subject":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests","message":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests\n\nWhen running Neutron extension tests and Tempest configured with\n\u0027use_dynamic_credentials \u003d false\u0027 the user_id and tenant_id are unset\nand results in the tests not being executed: \"Could not find user: None.\"\n\nBoth user_id and tenant_id are needed in rbac_utils [0]\n\n[0] https://github.com/openstack/patrole/blob/master/patrole_tempest_plugin/rbac_utils.py#L163\n\nChange-Id: I88d0624bc279b004ebe5f721cdef5adf12bd0ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2bd8e6f1ee4129ae2a38827993bbbad4051158b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/2bd8e6f1ee4129ae2a38827993bbbad4051158b4"}]},"branch":"refs/heads/master"},"4cd16e8fa13b6b69ca017b0fe9dd1bc7f5e9fc29":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-29 14:42:34.000000000","uploader":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"ref":"refs/changes/79/676979/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/79/676979/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/79/676979/6"}}},"commit":{"parents":[{"commit":"1099cdf21e6ce12b6d5e8684e08078339ff21181","subject":"fix: \u0027openstack-tox-docs\u0027 gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1099cdf21e6ce12b6d5e8684e08078339ff21181"}]}],"author":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-16 15:42:54.000000000","tz":-240},"committer":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-29 14:42:34.000000000","tz":0},"subject":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests","message":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests\n\nWhen running Neutron extension tests and Tempest configured with\n\u0027use_dynamic_credentials \u003d false\u0027 the user_id and tenant_id are unset\nand results in the tests not being executed: \"Could not find user: None.\"\n\nBoth user_id and tenant_id are needed in rbac_utils [0]\n\n[0] https://github.com/openstack/patrole/blob/master/patrole_tempest_plugin/rbac_utils.py#L163\n\nChange-Id: I88d0624bc279b004ebe5f721cdef5adf12bd0ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4cd16e8fa13b6b69ca017b0fe9dd1bc7f5e9fc29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/4cd16e8fa13b6b69ca017b0fe9dd1bc7f5e9fc29"}]},"branch":"refs/heads/master"},"e88d55248ad9aaa9876c906e533e0d456eb54520":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-09-04 18:52:19.000000000","uploader":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"ref":"refs/changes/79/676979/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/79/676979/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/79/676979/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/79/676979/7"}}},"commit":{"parents":[{"commit":"1099cdf21e6ce12b6d5e8684e08078339ff21181","subject":"fix: \u0027openstack-tox-docs\u0027 gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/1099cdf21e6ce12b6d5e8684e08078339ff21181"}]}],"author":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-08-16 15:42:54.000000000","tz":-240},"committer":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2019-09-04 18:52:15.000000000","tz":0},"subject":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests","message":"Fix \u0027use_dynamic_credentials \u003d false\u0027 with Neutron extension tests\n\nWhen running Neutron extension tests and Tempest configured with\n\u0027use_dynamic_credentials \u003d false\u0027 the user_id and tenant_id are unset\nand results in the tests not being executed: \"Could not find user: None.\"\n\nBoth user_id and tenant_id are needed in rbac_utils [0].\n\n[0] https://github.com/openstack/patrole/blob/master/patrole_tempest_plugin/rbac_utils.py#L163\n\nChange-Id: I88d0624bc279b004ebe5f721cdef5adf12bd0ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/e88d55248ad9aaa9876c906e533e0d456eb54520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/e88d55248ad9aaa9876c906e533e0d456eb54520"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
