)]}'
{"id":"openstack%2Fpatrole~682799","triplet_id":"openstack%2Fpatrole~master~I1337f3185d72108eef2968cca3cb81d52e742e90","project":"openstack/patrole","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"I1337f3185d72108eef2968cca3cb81d52e742e90","subject":"Add PDF building","status":"MERGED","created":"2019-09-18 06:16:39.000000000","updated":"2019-09-25 23:57:07.000000000","submitted":"2019-09-25 23:56:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682799-1569455763773-dadc74e4","meta_rev_id":"32e5e993b9d79a23ae79e651e5a0505e1d450f55","_number":682799,"virtual_id_number":682799,"owner":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 23:56:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2019-09-25 05:11:22.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-25 22:30:03.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-25 22:30:03.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 05:11:22.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-09-25 22:30:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-09-25 23:56:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0d582d771e2af1544eca94476a62a81258321bd","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-18 06:16:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"703e6ea121df8d9fc24972face378a9c8b32e859","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-18 07:25:22.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3786a43d316d1b7d4cec34b667231bc7eb884dc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:02:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/76a4c4c3c85046008624b437e1982558 : FAILURE in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ff9c99a16c542cc9b973eb41e0efae7 : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/330f5950481b4083aa5bc1d8e75c968d : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d25fc8086643aa9b4bfc88b98c5c06 : SUCCESS in 2m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1ee170d1e094739abae29970940c1aa : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa9fe94698444210ad07db9b78acb512 : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64cb928a06c84572aa86c9abe00d735e : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/667f77cf70ad4c1d915397c6ebdd99f5 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58ed12ba5d147d998e3145a3c9a92ec : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/065bce968fd244f394af5de4f88d8030 : SUCCESS in 3m 37s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/01d9f2f36dfa48a0962ab4e3b875ff0b : FAILURE in 1h 01m 33s\n- patrole-member https://zuul.opendev.org/t/openstack/build/ca4ab4d555814fb198429758825263b9 : FAILURE in 1h 14m 20s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/62388de20e184a9d8d896bfc9b902a90 : FAILURE in 1h 26m 05s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/131dbd4e5aa64e0a80f4a8ba57cca7f6 : SUCCESS in 1h 16m 09s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/1ffd15519d2b4802939df6e31ba7467f : SUCCESS in 59m 55s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/1b0c573aafcc47709fe1075e5b120604 : SUCCESS in 1h 13m 38s\n- patrole-member-pike https://zuul.opendev.org/t/openstack/build/abb2277a2ca544beac3ad2cc8155eb13 : FAILURE in 1h 09m 47s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/e96277492a5b470489c71a5dbcae4f45 : FAILURE in 1h 08m 53s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/70a83c791b264b719ade0026f5740117 : SUCCESS in 1h 16m 59s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/915e7830f1744b789eb07917a896c794 : SUCCESS in 39m 01s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/8bce3aaed45a47799bc92be3da7f3e98 : FAILURE in 40m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60404336d6525a57f89e4e11a2b0b94f182e15cb","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-09-19 17:51:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nThanks for taking this on. Just let a comment about why I think test-release-openstack is failing: \n\n2019-09-18 07:38:30.225328 | TASK [Run twine check on python dist tarball]\n2019-09-18 07:38:31.426317 | ubuntu-bionic | Checking dist/patrole-0.6.1.dev9-py2.py3-none-any.whl: FAILED\n2019-09-18 07:38:31.426567 | ubuntu-bionic |   `long_description` has syntax errors in markup and would not be rendered on PyPI.\n2019-09-18 07:38:31.426673 | ubuntu-bionic |     line 15: Error: Unknown target name: \"tempest\".\n2019-09-18 07:38:31.426805 | ubuntu-bionic |   warning: `long_description_content_type` missing.  defaulting to `text/x-rst`.\n2019-09-18 07:38:31.426898 | ubuntu-bionic | Checking dist/patrole-0.6.1.dev9.tar.gz: FAILED\n2019-09-18 07:38:31.427032 | ubuntu-bionic |   `long_description` has syntax errors in markup and would not be rendered on PyPI.\n2019-09-18 07:38:31.427125 | ubuntu-bionic |     line 15: Error: Unknown target name: \"tempest\".\n2019-09-18 07:38:31.427255 | ubuntu-bionic |   warning: `long_description_content_type` missing.  defaulting to `text/x-rst`.\n2019-09-18 07:38:31.896318 | ubuntu-bionic | ERROR\n2019-09-18 07:38:31.896736 | ubuntu-bionic | {\n2019-09-18 07:38:31.896864 | ubuntu-bionic |   \"delta\": \"0:00:00.701376\",\n2019-09-18 07:38:31.896970 | ubuntu-bionic |   \"end\": \"2019-09-18 07:38:31.461832\",\n2019-09-18 07:38:31.897072 | ubuntu-bionic |   \"msg\": \"non-zero return code\",\n2019-09-18 07:38:31.897170 | ubuntu-bionic |   \"rc\": 1,\n2019-09-18 07:38:31.897269 | ubuntu-bionic |   \"start\": \"2019-09-18 07:38:30.760456\"\n2019-09-18 07:38:31.897366 | ubuntu-bionic | }","accounts_in_message":[],"_revision_number":2},{"id":"8ca027b48a356082feaf2efe69cd3c4c128f3248","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-20 02:21:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18b588ac3b39a8826a96a0baa737f34660450b26","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-20 02:34:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f2e3383405ef19a19b579839fa8eb678e12bdd73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 04:14:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/048242f5a5554be081cace95bdc51c58 : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c796fc5b49c543ac88398493dd5eccc1 : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92d92512d6224a36a8baf5b5de548a73 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/511fd0316e624fe1a2f0a9f76df845eb : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6afdf8cb087496680ba218d8cdfe2d2 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082290d334b64ab7b8f964a0de700940 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac3785a51fa546248028b6965c9d949f : SUCCESS in 2m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d98f0e17345749b5a6860b3fe99ab545 : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f4d741c7734eb0a24880c33073bb47 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc98afd1809548f78c8cb53081efe0f5 : SUCCESS in 2m 57s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/f796bed0e87d42939e12ffc2ba9549c9 : SUCCESS in 57m 33s\n- patrole-member https://zuul.opendev.org/t/openstack/build/e436942a0b5341a6b22740ffa65dd277 : SUCCESS in 1h 13m 33s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/d5ee2c00f8ae4bafac6a5172f8b6b53f : SUCCESS in 1h 12m 32s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/3ab6d10d65364c26a55b26ea5373438f : SUCCESS in 1h 15m 50s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/e445bc8a8f1a49a2ae4962c2ec1c4f44 : SUCCESS in 1h 06m 08s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/c4bc351833f147788f265724e6d04416 : FAILURE in 1h 14m 53s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/6c8b94a3220f45a2a7e196e624b8311b : FAILURE in 1h 09m 21s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/0ebc876708b64fad9acf28455cc68665 : SUCCESS in 1h 08m 58s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/76d506c5f121424798dfa2eb4819f661 : SUCCESS in 37m 24s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/c95e3858dd784dcca052d7cec1df2f88 : FAILURE in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72ecf1f73f7316a79f5e8a1f6ae6751d15cb7c24","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-09-20 18:09:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"13cec45ddf94c7227651c69c3788a3befd9f60c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 19:44:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc5ab24fb64c4a25ab937f27eda24352 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d0505c2fa9c4d20bac191f7e2c367d1 : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f5dda248284529b365e0cd41a919d4 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/167f5048db0a445aa777091899f44427 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce901db28b074fc78c4a2258094c9f2b : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/089bb144d10347f994d692076473a722 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fe5df70d91a4665afd16004771c7528 : SUCCESS in 5m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ab46966b45094b23a63d3e48c2052398 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f445077f393145f7968f55932fbc049d : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/823552e1dabb4eb6b9aaf9af9e61f281 : SUCCESS in 3m 34s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/595ab54630024137b7744b98d06ebc73 : SUCCESS in 1h 05m 36s\n- patrole-member https://zuul.opendev.org/t/openstack/build/54e84222f5be426a876cbe059e6c0809 : SUCCESS in 1h 14m 40s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/17c1a0a32791401685985fad208c1e51 : SUCCESS in 1h 12m 12s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/d3ab833029d8498eaf81fb0e25bb6446 : SUCCESS in 1h 21m 32s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/71f0d638b7d748fb8620f2cef13e5c58 : SUCCESS in 1h 18m 13s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/320ee511562140bf867d593e138b0c02 : SUCCESS in 1h 07m 55s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/7819bc45df5042ac832203ca8c36e4d9 : FAILURE in 1h 11m 59s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/45266d79a9eb49df8b4495e727cc07d8 : SUCCESS in 1h 23m 34s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/40ea709a13ca47a591fb511e1e343b0b : SUCCESS in 44m 26s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/261a6f340eff40a0a97cf81aa04c0777 : FAILURE in 44m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c1d450a040d015a650a8ea83373ddfa17fed7bf1","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-09-22 05:05:38.000000000","message":"Patch Set 3:\n\nhttps://storage.bhs1.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_f44/682799/3/check/openstack-tox-docs/f445077/docs/ and https://storage.bhs1.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_f44/682799/3/check/openstack-tox-docs/f445077/pdf/doc-patrole.pdf both LGTM. Only question I have is why the README is trimmed down so much? I see that happened to Tempest\u0027s README as well: https://github.com/openstack/tempest","accounts_in_message":[],"_revision_number":3},{"id":"048ba1118ec5e794d82887e3c097685cf3d23964","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-24 01:45:52.000000000","message":"Patch Set 3:\n\n\u003e Only question I have is why the README is trimmed down so much? I see that happened to Tempest\u0027s README as well: https://github.com/openstack/tempest\n\nI feel the current README is a bit long as I mentioned about it in the commit message. And, some other project\u0027s (such as nova, glance, cinder, keystone, neutron) readmes are short like below.\n\n\n- https://opendev.org/openstack/nova\n- https://opendev.org/openstack/glance\n- https://opendev.org/openstack/cinder\n- https://opendev.org/openstack/keystone\n- https://opendev.org/openstack/neutron","accounts_in_message":[],"_revision_number":3},{"id":"5ad0848845a85abf82aff6d9b26fea279211d8a3","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-24 06:38:52.000000000","message":"Patch Set 3:\n\nMasayuki Igawa wrote:\n\n\u003e \u003e Only question I have is why the README is trimmed down so much? I see that happened to Tempest\u0027s README as well: https://github.com/openstack/tempest\n\u003e\n\u003e I feel the current README is a bit long as I mentioned about it in\n\u003e the commit message. And, some other project\u0027s (such as nova, glance,\n\u003e cinder, keystone, neutron) readmes are short like below.\n\u003e\n\u003e - https://opendev.org/openstack/nova -\n\u003e https://opendev.org/openstack/glance -\n\u003e https://opendev.org/openstack/cinder -\n\u003e https://opendev.org/openstack/keystone -\n\u003e https://opendev.org/openstack/neutron\n\nBut I also think this is a different topic. So, I\u0027ll update my patch set to recover the trimming. That should be easier to review.","accounts_in_message":[],"_revision_number":3},{"id":"c5750104e704fcaecb60b3433d995823eed34755","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-24 06:39:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f77fdce86881ee4c295d4f8711ef9d878db66565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:07:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9b621a68f2544583a20c55bb457d00d7 : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c294fd1d4d754583900bb7e3f7c3f66a : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c88ad87043ff4cb6b8d07377023ce637 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e7e54e1d5704fb6ac70be7eb5943bc6 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/831c944557fb4ee791c2a3b112f1858e : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03273c85c8af4829b4ac1f88e2fc0630 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b70a14d9e4584009b63ee6bbab676dc1 : SUCCESS in 13m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/98943476890d4171bb0be7c60d66166f : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8361ba3da13349ed8be6d632bd5c0822 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4845d140914433ca135d7a2a50f9822 : SUCCESS in 4m 40s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/66f6b1f6d75e4937bc1ca857fae01d61 : SUCCESS in 57m 36s\n- patrole-member https://zuul.opendev.org/t/openstack/build/2c23aa0043cf4e23adddee879b02e354 : SUCCESS in 1h 14m 49s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/8e5a77089fc843329232b05f41303baf : SUCCESS in 1h 11m 42s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/413d1e611e494c2bbc531603b3f479c9 : SUCCESS in 1h 01m 25s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/ed2f9bc49a834531bef9a3b10f26e913 : SUCCESS in 1h 05m 10s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/edd3b9b74864498dbf558cc15ec70438 : SUCCESS in 1h 19m 38s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/d66566060df04da2822b26e608d3b144 : FAILURE in 1h 17m 09s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/20d94f1ff46a4511afa6c6d6a94a9966 : FAILURE in 1h 05m 09s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/319e01221c574c48ab3d70dafb2fbc4e : SUCCESS in 43m 30s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/a460e8ae35fb41bfa07d04dec13957ab : SUCCESS in 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f8130997f4f544c16ce86bdbdd11c4204b6e3416","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-09-25 05:11:22.000000000","message":"Patch Set 4: Code-Review+2\n\nNice work, LGTM.","accounts_in_message":[],"_revision_number":4},{"id":"02b5af29c9963c0d3c27beea7ea08fa41ec15323","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-25 22:30:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":4},{"id":"1e07e267fd1c58870c9428be44317c5483b223f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 22:30:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c4f6b8fb66b9db9d97bb57557cf2238b33df0e72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 23:56:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"eee9f01b184d455f6571cd6c3aa375e3fe9ffca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 23:56:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f0ea9dbd6cf47cfb1779a1b9a8c1224 : SUCCESS in 2m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a182fa66ca0b4524a1c69360603f462e : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a2b1b96e311486893393e4b68ed2fae : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3b6089310834c049019d7aef7ce9c7e : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d5b11377df4572ae8cccf41919c73a : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ab0189a4965407cba9c0a7b99f652c4 : SUCCESS in 5m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c222288782bf43b7a861a3559594f9be : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/babbd66843604c7ea5fc91ec6be12ae6 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3ca88dc8d7a483a8c9a3b711032567f : SUCCESS in 3m 32s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/c60764cbafc446e5b260e4802d0e4052 : SUCCESS in 1h 10m 40s\n- patrole-member https://zuul.opendev.org/t/openstack/build/15ec6bd6d7314d97bce6ef8713def03d : SUCCESS in 1h 20m 26s","accounts_in_message":[],"_revision_number":4},{"id":"2e3bbcf95b319aaaf6080ba535036bad2d12ff02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 23:57:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2706c1ffc3bb402c85ea7eeb598cf332 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a095fd882c74b2bb0e3ccf933fe5d2f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"94e6b1875a89ad308a4fffba479aa9d0abaeeaaa","revisions":{"f70bd3674735153e04ef49f8840b6df1c437c3ab":{"kind":"REWORK","_number":1,"created":"2019-09-18 06:16:39.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/99/682799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/99/682799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/99/682799/1"}}},"commit":{"parents":[{"commit":"fe4e5fd7b5ee9a511d95e88fd124b710dcc341ba","subject":"Merge \"[doc] Fix docs job by adding _static folder to avoid Sphinx warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fe4e5fd7b5ee9a511d95e88fd124b710dcc341ba"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-09-18 06:08:56.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-09-18 06:16:23.000000000","tz":540},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit add PDF building tox.ini environment and options for LaTeX\noutput. overview.rst is copied from README.rst. And, the original\nREADME.rst file is shrunk because we don\u0027t need such a long information\nin the README file. People can see the same contents in overview.rst\nnow.\n\nChange-Id: I1337f3185d72108eef2968cca3cb81d52e742e90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f70bd3674735153e04ef49f8840b6df1c437c3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f70bd3674735153e04ef49f8840b6df1c437c3ab"}]},"branch":"refs/heads/master"},"5cb31affe6556162935cdb979717864f30398d94":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-18 07:25:22.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/99/682799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/99/682799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/99/682799/2"}}},"commit":{"parents":[{"commit":"fe4e5fd7b5ee9a511d95e88fd124b710dcc341ba","subject":"Merge \"[doc] Fix docs job by adding _static folder to avoid Sphinx warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fe4e5fd7b5ee9a511d95e88fd124b710dcc341ba"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-09-18 06:08:56.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-09-18 07:24:58.000000000","tz":0},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit add PDF building tox.ini environment and options for LaTeX\noutput. overview.rst is copied from README.rst. And, the original\nREADME.rst file is shrunk because we don\u0027t need such a long information\nin the README file. People can see the same contents in overview.rst\nnow.\n\nChange-Id: I1337f3185d72108eef2968cca3cb81d52e742e90\nStory: #2006070\nTask: #35469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5cb31affe6556162935cdb979717864f30398d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/5cb31affe6556162935cdb979717864f30398d94"}]},"branch":"refs/heads/master"},"ee624cd67fbf08f50d0cc69c4f249bdb43315d9d":{"kind":"REWORK","_number":3,"created":"2019-09-20 02:34:12.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/99/682799/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/99/682799/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/99/682799/3"}}},"commit":{"parents":[{"commit":"fe4e5fd7b5ee9a511d95e88fd124b710dcc341ba","subject":"Merge \"[doc] Fix docs job by adding _static folder to avoid Sphinx warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fe4e5fd7b5ee9a511d95e88fd124b710dcc341ba"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-09-18 06:08:56.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-09-20 02:33:56.000000000","tz":540},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit add PDF building tox.ini environment and options for LaTeX\noutput. overview.rst is copied from README.rst. And, the original\nREADME.rst file is shrunk because we don\u0027t need such a long information\nin the README file. People can see the same contents in overview.rst\nnow.\n\nChange-Id: I1337f3185d72108eef2968cca3cb81d52e742e90\nStory: #2006070\nTask: #35469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ee624cd67fbf08f50d0cc69c4f249bdb43315d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/ee624cd67fbf08f50d0cc69c4f249bdb43315d9d"}]},"branch":"refs/heads/master"},"94e6b1875a89ad308a4fffba479aa9d0abaeeaaa":{"kind":"REWORK","_number":4,"created":"2019-09-24 06:39:25.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/99/682799/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/99/682799/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/99/682799/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/99/682799/4"}}},"commit":{"parents":[{"commit":"fe4e5fd7b5ee9a511d95e88fd124b710dcc341ba","subject":"Merge \"[doc] Fix docs job by adding _static folder to avoid Sphinx warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/fe4e5fd7b5ee9a511d95e88fd124b710dcc341ba"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-09-18 06:08:56.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-09-24 06:32:25.000000000","tz":540},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit add PDF building tox.ini environment and options for LaTeX\noutput. The chapter header in README.rst has been changed because \"Team\nand repository tags\" is not appropriate for the title of this document.\nAnd, this commit also updates repository URLs to opendev.org.\n\nChange-Id: I1337f3185d72108eef2968cca3cb81d52e742e90\nStory: #2006070\nTask: #35469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/94e6b1875a89ad308a4fffba479aa9d0abaeeaaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/94e6b1875a89ad308a4fffba479aa9d0abaeeaaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
