)]}'
{"id":"openstack%2Fpatrole~689052","triplet_id":"openstack%2Fpatrole~master~Ibe3ddbb18c289662940f442d6d75236f04a6b2cb","project":"openstack/patrole","branch":"master","topic":"bug/1843290","hashtags":[],"change_id":"Ibe3ddbb18c289662940f442d6d75236f04a6b2cb","subject":"Unskip test that relied on Neutron bug being fixed","status":"MERGED","created":"2019-10-16 21:48:38.000000000","updated":"2020-03-28 02:37:18.000000000","submitted":"2020-03-28 02:35:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689052-1585362948700-68d14c44","meta_rev_id":"3b33731229d44948d94aa7f7997e5e2265caacb8","_number":689052,"virtual_id_number":689052,"owner":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-28 02:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":1,"date":"2020-03-21 23:31:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":1,"date":"2020-03-21 23:31:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2020-03-21 23:31:46.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":2,"date":"2020-03-21 23:31:46.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":2,"date":"2020-03-21 23:31:46.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":1,"date":"2020-03-28 00:07:35.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"value":0,"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"value":0,"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 21:50:06.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2019-10-16 21:50:06.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2019-10-16 21:50:06.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-10-16 21:50:06.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-10-16 21:50:06.000000000","updated_by":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"reviewer":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"state":"REVIEWER"},{"updated":"2019-10-16 21:50:06.000000000","updated_by":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"reviewer":{"_account_id":8911,"name":"Sergey Vilgelm","email":"sergey@vilgelm.info","username":"sergey.vilgelm"},"state":"REVIEWER"},{"updated":"2019-10-16 21:50:06.000000000","updated_by":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"reviewer":{"_account_id":16274,"name":"Mykola Yakovliev","email":"VegasQ@gmail.com","username":"vegasq"},"state":"REVIEWER"},{"updated":"2019-10-16 21:50:06.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2019-10-16 21:50:06.000000000","updated_by":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"reviewer":{"_account_id":23185,"name":"Samantha Blanco","email":"samantha.blanco@att.com","username":"sblanco"},"state":"REVIEWER"},{"updated":"2019-10-17 07:03:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-17 18:19:19.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-10-31 10:10:36.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2019-11-14 05:43:58.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2020-02-15 02:19:34.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2020-03-28 00:07:35.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-28 02:35:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4402ec7897834bdc64a3b9eceb8b4e1ef38c802f","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-10-16 21:48:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df6cf80e76698856748a52bf3686d9a575454458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 23:20:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1856d07ef3d46c2ba248d0e66ce4d0e : SUCCESS in 3m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a9ec25a2c14442eac7bef4dd565326c : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fa00c30693a46fb9ef53bbae980de10 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caa00df963624f7696b8b71ad6a8dc8c : POST_FAILURE in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1383853eeb44e6fb1eeff2e985cf1b9 : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/257202157b434967ad66d601049286de : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d63dcf58c5f40bc84bbf8c7f7a53fb8 : SUCCESS in 6m 35s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/d48acd4ac02245fda7d141a5e7ae5bbf : SUCCESS in 1h 06m 55s\n- patrole-member https://zuul.opendev.org/t/openstack/build/4fb4f28c48134cfb8b13ecc8b6111d78 : SUCCESS in 1h 23m 32s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/d434db4d586d4877b5b4fa2770295d87 : SUCCESS in 1h 12m 11s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/0c14e4fb368f403dbd3a701e24cfd292 : SUCCESS in 1h 05m 34s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/fce653f001844bd49a9c35bd651012aa : SUCCESS in 1h 19m 16s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/d63a47be44c34f2990db5dd44d4db90c : SUCCESS in 1h 12m 25s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c77b2bb1cc9845cc9efd9f52ea55cee1 : SUCCESS in 1h 05m 36s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/48dc9b5411074b1994fd5663706a9a17 : SUCCESS in 1h 07m 38s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/806c3949ecfb48d4b4c021c61707ebcf : FAILURE in 34m 55s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/46633cf8dcd846f897df09d653e3319a : FAILURE in 38m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab0df78e7815c84a21c63227a0a3b11cda5eb85f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-17 07:03:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d95de630af5071c22622aad5d5d758925252422","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-10-17 16:48:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75f7db41618fc43a61e526fe1872e34bb7a62f5c","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-10-17 18:19:19.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM. There is an unrelated failure here: https://1d3288128fd897bdd329-62f6dec4e35d4346ff015b62b238dd72.ssl.cf5.rackcdn.com/689052/1/check/patrole-extension-admin/806c394/testr_results.html.gz\n\nThat is due to the existing logic expecting a testtools.TestCase.skipException I believe but with the recent Tempest change a unittest2.case.SkipTest is being raised instead.","accounts_in_message":[],"_revision_number":1},{"id":"90bb334f94e6c2267e9bb9288761d25521287bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 18:21:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fea23968d1364ac2be1a7250e2fcc275 : SUCCESS in 3m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0f69b83383a4d858e20f6a021908e86 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c28ed63394334e0d876c480375ad4b47 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9509bafaa244577877cd03a968996d2 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdd6d27017894e6b9f139710ac6cdc22 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7567209b38a749e2835911f09bde3bd1 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c083f8b5e1f4bc5ad302a258c3662e1 : SUCCESS in 5m 33s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/0306710daad049bd92e7191a68fda5c2 : SUCCESS in 57m 01s\n- patrole-member https://zuul.opendev.org/t/openstack/build/6453f1bdc8cd4e13a04822b7439dcf15 : SUCCESS in 1h 15m 32s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/fa81b36dbbc54e1a9152faff61c78988 : SUCCESS in 1h 30m 08s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/fad3b4e7654b49238675faa61079acbf : SUCCESS in 1h 19m 48s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/0547c53ea02e43cb891c7ebddc048873 : SUCCESS in 1h 17m 57s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/9f716965187b418eb816113760759af1 : SUCCESS in 1h 03m 47s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/c7825e6d08174c748047aa89a8708c57 : SUCCESS in 57m 00s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/9357c08f7b994f56b7f1b2a1ee6aa8ce : SUCCESS in 1h 20m 55s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/7b8474aa488c4f60b40d0830b4043b45 : FAILURE in 39m 02s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/40bf28daeb6741818edca5dd850e9656 : FAILURE in 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"015e9d12b6b39737b887cc9702042d453bad9c54","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2019-10-31 10:10:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0984a040e60ed3d6d5df97013edc09b4a35821be","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-11-14 05:43:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f9c82cafc08b03c8800b0f08a293195ae25e58dd","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2020-01-02 15:18:36.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad784bef20818b657c5792193e86b8fbef2f7297","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 15:18:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9add2824a81b5bed6e2e2e3b7092b5471fad9460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 16:38:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ba0b6f2e1e2414b81fa46bd33e00e74 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f48ab1df6b324794ba7d23ec40e1a97b : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34e4e76fbfb74513b9a3135c554d38d4 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb137c1c229b42c4bad8972403aabdad : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eff4c5c54f464d43b0dae1a5f14e05ce : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff4ba3d8ca64b66887f31090c66d129 : SUCCESS in 7m 44s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/2cedfe085c7944b7a6632cf51461a633 : FAILURE in 1h 06m 48s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c6972408aa7c458c99eb38761878f7eb : FAILURE in 1h 16m 23s","accounts_in_message":[],"_revision_number":1},{"id":"408dd4eb182f624ffb590297863add6c66fbb4be","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-10 21:01:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a9795d0279095298a24930f0d6a7d5d6db43f03d","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2020-01-10 21:03:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"78d0c5bffcfc2d1d0d04c1b2fc976e99ff08cad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 22:21:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8f71cc094b44ae58e015dd17313ff99 : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e64cc946a94747c0af5de13f413df30a : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5142e12227e4b3286f2cfdb7549b9ac : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e5a4c46dddb4707ba30c378858e9013 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29dfbbb013f4707a1ffb787a2e33597 : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6822df372fee4546bc42a6166699c8fa : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aac170bd9f644f08f603236162fe199 : SUCCESS in 5m 02s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/44220c4b1b124ebdaf77d7f82b175d27 : FAILURE in 52m 34s\n- patrole-member https://zuul.opendev.org/t/openstack/build/c2590e70cada4544a23642ff1c0d324e : SUCCESS in 1h 05m 30s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/fba613e55d2d4f4c9c1f88ce6ca44cf5 : SUCCESS in 1h 12m 29s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/634659b936ef404a9975907777c28c11 : SUCCESS in 1h 12m 19s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/deecbc8f624543f7bce6d17d7c805a74 : SUCCESS in 1h 09m 10s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/7ebd9ff5c1f9419a9cc4bdd978b423b3 : FAILURE in 1h 09m 13s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/11344e26a14d40aa9f153e2223d074e0 : SUCCESS in 1h 01m 16s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/6b93dde2ec944da3acd4c6198b052bc4 : SUCCESS in 1h 00m 49s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/3a9437d982aa4f918671709baa8b13c7 : FAILURE in 34m 57s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/8e54c466ffeb4b319c2f03e9412919a3 : FAILURE in 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8e344f5ba9847e797f1064b81ca50f360989b38","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-01-12 20:07:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5de30e061690ffacb785175b3755bb95f0512664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 20:27:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f9c61c3471f4240bdcad5efd045770b : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95f84bd67e404cf2ad3c6103f7a7920e : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e01698f0b91545d99c4315ecbd13138c : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/168fdfeb45bb41beb7286eb6b71b2813 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a256d94650e49bba810dae1e013931c : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a004f36e93745459c6723eed6661794 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df20782b6b54bc5ba05656dbb34cbd6 : SUCCESS in 6m 24s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/25e05bddde3b4555985ac14612aff8f6 : FAILURE in 9m 13s\n- patrole-member https://zuul.opendev.org/t/openstack/build/b51ada2e693642c991878109ef2510d2 : FAILURE in 9m 03s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/81a01558cb2a4a89a6bb3f35f0c793d7 : FAILURE in 9m 06s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/68f01b32417a43b9899cdb863f58eaed : FAILURE in 9m 24s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/9355439faaeb4d5a9ddab9a8c2f81428 : FAILURE in 8m 43s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/61644a8286d64aec980c8874d4f79080 : FAILURE in 7m 55s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/91286b49bd9d41ee8bc7776de0742b1c : FAILURE in 12m 48s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/3ca46c31458442c68e3ddceb464b2401 : FAILURE in 11m 49s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/2b66de56003442229db386a745cc8e1d : FAILURE in 8m 39s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/0d84e4e9975b4b0aa7b653a07163b814 : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64891ad9e10f88950f6163936dadf5231bd6b7a1","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-01-22 02:57:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"337811c0db5c74edae5737d88b64f9ee885665fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 04:36:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed7f6b396c034ab8a20c92043f575c47 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49074102b8ed4bb7bb641290f353e4ad : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a620bb15cab4109b095bc594f5441c6 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/060af6474f3344e9a16b37700e454312 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0751a20c1ca45eeaf4b2b934478a6e7 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cdb53bafabbd4888af58e3b0c7b068bd : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddab567ade464cc182465ea313ed4cc4 : SUCCESS in 5m 39s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/785beef01f5d469ba7912dea07bb2a4e : SUCCESS in 56m 21s\n- patrole-member https://zuul.opendev.org/t/openstack/build/aa237459190d4a2f9a4d52ee4e19619d : SUCCESS in 1h 05m 10s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c44efa1d2340468cad56aaecca03b325 : SUCCESS in 1h 24m 49s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/bf6261a2a1f94182af6a176d9664896b : SUCCESS in 1h 21m 34s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/c9b46b9b3c804d9dbaed1030a1b8543d : FAILURE in 1h 32m 58s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/be9012d041cb43a9b41138c737faa16a : SUCCESS in 1h 19m 22s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/ee7a6af9e90d42db9a8ab0ec6834db29 : SUCCESS in 1h 34m 26s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/d98e3b9dd2154b4f97b19d523011a1a6 : SUCCESS in 1h 02m 07s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/7fe8d8df14c34436a3381976cdc4d85d : FAILURE in 1h 05m 02s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/de27e2f2b9344f33a0ee8d0a939e74f4 : FAILURE in 33m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3bb67a6fb8f8256ba5d1f298db5d7e4437797aa7","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-02-11 12:58:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9123c2683354fc2b2e6648342b5e0b970bb14c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 14:09:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d36661974b8c4ec2ba34f2780ffee966 : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b7f2105d0444b3c8d938958e772f25a : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a37ee4bd5e84ffc9e41b6c56487e964 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/397fa17951854cc8991b6043e3f6fe27 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62f79c2999f34e6192e523ec080c879c : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7495d11758040faa375acc6701feb09 : FAILURE in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b4cf5a9cde94b90a8bd6109f70322e3 : SUCCESS in 6m 07s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/f5316a85d75c4da28ac478df85b665ea : SUCCESS in 49m 01s\n- patrole-member https://zuul.opendev.org/t/openstack/build/88f734f4f8964380bc4c4d6dd6f1ca4a : SUCCESS in 1h 05m 51s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/636a2bccc1894a33ab417d640bc75590 : SUCCESS in 1h 09m 28s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/893f36632566438395dddd1668b94c90 : FAILURE in 35m 30s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/07dd5d25f5a243dc93bd2c6f1529e105 : FAILURE in 11m 50s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/c95b6a76cf644eae92cd3a3808c8049b : FAILURE in 11m 49s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/a361d644ec104bbda12370c127892731 : SUCCESS in 1h 02m 33s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/7281d618efca4d6abb3bb14bbe6094ca : SUCCESS in 1h 00m 56s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/aa7973b6d0564eaba304d8516109154e : FAILURE in 47m 14s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/fd132dc2d3084e77b10d5810fc20d9e1 : FAILURE in 34m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ffaba917373eeffbfd592b6a6d2e46f618910f86","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-02-15 02:19:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fcdddb035725884b7c481f8dcd46317d51298642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 03:43:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/088ccca13195496a91d6ee6c0cac89ce : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8194e8eabddb42178e237a59cf9b5f05 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c441eaec1995431993cd8dd2017002c0 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae057638609a424582738c66d1c0f0e7 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a3bbdb182cc44e1a9df4178c823a199 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c4a5a3bb3504d13988f3e9d82cebfa7 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3687ad109cb242328299a9c28412508b : SUCCESS in 6m 36s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/beada5bb8b7d47cdbbdb3300f8db48b5 : SUCCESS in 1h 03m 11s\n- patrole-member https://zuul.opendev.org/t/openstack/build/3dee22b2bb6849339203ea100d878287 : SUCCESS in 1h 21m 35s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/3d490f67cd8249c2bcc6295e2cba5c2d : SUCCESS in 1h 16m 53s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/b1c1e06cfd6246749695db704b119ac5 : FAILURE in 25m 16s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/1213675fce624c8fa7c6200af39e493e : SUCCESS in 1h 02m 11s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/d58a5a829bcc4e9d90e699996c9b4280 : FAILURE in 32m 55s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/d144d800b20d4cf98a54a212294d236d : SUCCESS in 59m 10s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/84fe1446f0cf457e8b9e6ade8d718acc : SUCCESS in 1h 08m 38s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/32eb1b47c76f4b5e839da87bac1cd477 : FAILURE in 42m 44s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/bd65f4667ad84accb4ea3f6467a9964a : FAILURE in 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d35e7bb7a040ea4182306357433f20b016c51eab","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-05 01:32:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"333385deb5913c57ff8d4ec663bbb642ba4d0ee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 02:47:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/954092b4de8c4de39b83196603c777bb : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c61e4de5da734655a72683da9304a02c : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e8628ae4414ce4a276a1a220bf41a3 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04f2ee9f6ccf439b90467fb5c3730bd4 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/952526b2870a4c39865f1df84515cae9 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfcb4c765bb74ae582b5368decc96286 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea7db2082f74ffe908fb7dedb461952 : SUCCESS in 5m 48s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/5fa1f6a76a064bc58c8f84de3603e87d : SUCCESS in 51m 45s\n- patrole-member https://zuul.opendev.org/t/openstack/build/b567f59d9cc84a22933db0aec1570909 : SUCCESS in 1h 08m 19s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/c2402a2846fd4e678fe745668122808b : SUCCESS in 1h 09m 24s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/954fe2b3c692460d8d7bb0536b722787 : SUCCESS in 1h 07m 08s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/63158697023a46859819b1196a10b3ef : SUCCESS in 1h 09m 19s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/a0a2bc55d8c549a997f5d6d57fc16540 : SUCCESS in 1h 06m 07s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/58919535f4b944d2a2f99541e35f1b71 : SUCCESS in 1h 01m 44s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/fc7efd67296e420f8b18702422db14dd : FAILURE in 39m 18s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/3e12184743e643709decc24993622e7d : FAILURE in 36m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e625dc7e9996b1d58e49d094e79d9dee202e946","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-15 18:25:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cf95c7366123c8c68864ad31a0105d5de97f65db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 18:25:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"514842eb12c145c8cacf8c925f55894c62083ce5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 19:39:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15dabdde38934891bdef59def04146a1 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4258d24d5d964808a664c61331e8da01 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/707b5537abd040b996dca6aaadcd2779 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d2d64a16a374719a42702fe9fb0ff3f : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dec195aac704fc482d9f7053317f03f : SUCCESS in 5m 46s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/201464d4deaa4338acbb11cab8b4eec0 : FAILURE in 1h 02m 12s\n- patrole-member https://zuul.opendev.org/t/openstack/build/d620f9c42998482fb69c995d9a3e864f : FAILURE in 1h 12m 14s","accounts_in_message":[],"_revision_number":2},{"id":"9c88e2de7c4e493aed6571dc6036cfc9b9a0ac3f","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-03-21 23:31:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d6a8b0db2a9e52d35b6bb7dab83e585206a56124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 00:45:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aecc1d13b5144c3a8877c43ed7621733 : SUCCESS in 2m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/518c3a3fa36541b2b8f5d343956d9dd0 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c18a131dbc246529ec6a0f120eb9fdd : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4734eb6f35354cf5af6a33153d49a7a3 : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/310916d7d61548a3805bcc9d41e87099 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9908f9bd8cac43ddbdf6068cc45c557c : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50704b5bd65642b594a877a2e99fc8d7 : SUCCESS in 5m 36s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/6dbcd9b03a184c0189e05ca45a6abba4 : FAILURE in 54m 43s\n- patrole-member https://zuul.opendev.org/t/openstack/build/6316d79bd24b40b38df1a41c9b17b7ea : FAILURE in 1h 05m 10s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/60677d27fa6e4692a57d2e81f3082078 : FAILURE in 1h 02m 15s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/e6710ccfc74d42e5a5099fecc402442b : SUCCESS in 1h 06m 51s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/125209501640410f8af1e534345b7885 : SUCCESS in 1h 11m 23s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/87c350a96c5243d9bbda223166c13561 : SUCCESS in 57m 06s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/3f32375998a8426b884e6ee5671f3786 : SUCCESS in 59m 33s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/d75f459405bc4299bd2cf39faace7cf3 : FAILURE in 33m 42s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/c002050dedd144ff933e72f8decfa442 : FAILURE in 35m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4ac7c2f94d818baf3e83c3d47fa2b187e0198f37","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-28 00:07:32.000000000","message":"Patch Set 3:\n\ngate should be happy now. fixes are merged,","accounts_in_message":[],"_revision_number":3},{"id":"98bd6831a0355369e8bcd859c67875ab53f2525e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-28 00:07:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c80c317102b5fdc78375325b6e72e296d0b05e6d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-28 00:07:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1cb08973ec4332dfdac60bac2817bf22b09122eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 01:25:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e61d686c132421ba0261a4ae7c711a6 : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fb9475975ec4ed19f4c3b19e2cb9829 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd9e86e1ded4799b24af3cbdac96ec4 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d227f575548e4b66b68a5425f1c91e88 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/876057ce75d24f3d961913b8492f2a59 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15a88426cd5b4bafb86837ffc17f89c9 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcff7b6ec4474b8592ba552114d2098e : SUCCESS in 5m 37s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/27bd78507a3e4a68b9171558b55e1771 : SUCCESS in 55m 42s\n- patrole-member https://zuul.opendev.org/t/openstack/build/8db201b3a5ca48b0887ad6a5926c21b7 : SUCCESS in 1h 04m 11s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/0d263ebdc9934e12b8d79ea78785b382 : SUCCESS in 1h 04m 22s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/4bad2592a03b4476a90c5ce32922e823 : SUCCESS in 1h 15m 06s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/e9c8b5446e114b548c07114659bf3777 : SUCCESS in 1h 10m 08s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/2313d5833d3c499aafe6793f9746e42a : SUCCESS in 59m 50s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/9e012b875bf144cf99f8964912b5c0e7 : SUCCESS in 59m 35s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/9a3cfb6f503844d2a4bd7df185bf9b06 : FAILURE in 35m 37s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/b453dd95294f4a60b91dac8c0fec0662 : FAILURE in 35m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d42b97f50701242abf153efe6100b96f40af0278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 01:25:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6bea1cf29242760ac11211a147e013b086bb0976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 02:35:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad4b893df78942f79c56e63db30a1f9e : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/952f6ab5543e4e6a9ed993788bfdd9a9 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5eac17914004c469911db3bf407c96e : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e020e2a96fc24a43bd3aa91d1649ec46 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c060aed2e8a43ad973ad2f6279bf260 : SUCCESS in 5m 33s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/599ff848f6904d26882d7863be29e061 : SUCCESS in 53m 58s\n- patrole-member https://zuul.opendev.org/t/openstack/build/1c4b101810ee43e0b95b702b944935ad : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":3},{"id":"9c0660de3f8a5923f80fa2a855a7fe9b41e7d59b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 02:35:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"64e455701067aeb3b9cb4784b94d7f99d3bfdbbd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 02:37:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b31a1701e145a1ab8e83766c73aa03 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"595e61d17cb27900546bdfc022121cb816283f6f","revisions":{"25f32c7ec91c9479c6bf52f149ea397e83c5542d":{"kind":"REWORK","_number":1,"created":"2019-10-16 21:48:38.000000000","uploader":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"ref":"refs/changes/52/689052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/52/689052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/52/689052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/52/689052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/52/689052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/52/689052/1"}}},"commit":{"parents":[{"commit":"45f01b741754f91023f7038851fc018c8fbd538d","subject":"Merge \"Fix missing \"raise\" before cls.skipException\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/45f01b741754f91023f7038851fc018c8fbd538d"}]}],"author":{"name":"Rick Bartra","email":"rb560u@att.com","date":"2019-10-16 21:46:49.000000000","tz":-240},"committer":{"name":"Rick Bartra","email":"rb560u@att.com","date":"2019-10-16 21:46:49.000000000","tz":-240},"subject":"Unskip test that relied on Neutron bug being fixed","message":"Unskip test that relied on Neutron bug being fixed\n\n\u0027test_delete_flavor_service_profile\u0027 no longer has to be skipped as\nLaunchpad bug 1843290 is now fixed.\n\nChange-Id: Ibe3ddbb18c289662940f442d6d75236f04a6b2cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/25f32c7ec91c9479c6bf52f149ea397e83c5542d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/25f32c7ec91c9479c6bf52f149ea397e83c5542d"}]},"branch":"refs/heads/master"},"107f987b8ea7ec1fced86f989312bf097b809bc6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-10 21:03:11.000000000","uploader":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"ref":"refs/changes/52/689052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/52/689052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/52/689052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/52/689052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/52/689052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/52/689052/2"}}},"commit":{"parents":[{"commit":"f8923d1ddfd2608a5f402c7014f9e95321c43caa","subject":"fix: admin and member gates are broken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f8923d1ddfd2608a5f402c7014f9e95321c43caa"}]}],"author":{"name":"Rick Bartra","email":"rb560u@att.com","date":"2019-10-16 21:46:49.000000000","tz":-240},"committer":{"name":"Rick Bartra","email":"rick.bartra@att.com","date":"2020-01-10 21:03:11.000000000","tz":0},"subject":"Unskip test that relied on Neutron bug being fixed","message":"Unskip test that relied on Neutron bug being fixed\n\n\u0027test_delete_flavor_service_profile\u0027 no longer has to be skipped as\nLaunchpad bug 1843290 is now fixed.\n\nChange-Id: Ibe3ddbb18c289662940f442d6d75236f04a6b2cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/107f987b8ea7ec1fced86f989312bf097b809bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/107f987b8ea7ec1fced86f989312bf097b809bc6"}]},"branch":"refs/heads/master"},"595e61d17cb27900546bdfc022121cb816283f6f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-21 23:31:46.000000000","uploader":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"ref":"refs/changes/52/689052/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/52/689052/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/52/689052/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/52/689052/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/52/689052/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/52/689052/3"}}},"commit":{"parents":[{"commit":"f75396ceac7f073f455701e7b7d3d335f8d73a3c","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/f75396ceac7f073f455701e7b7d3d335f8d73a3c"}]}],"author":{"name":"Rick Bartra","email":"rb560u@att.com","date":"2019-10-16 21:46:49.000000000","tz":-240},"committer":{"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","date":"2020-03-21 23:31:46.000000000","tz":0},"subject":"Unskip test that relied on Neutron bug being fixed","message":"Unskip test that relied on Neutron bug being fixed\n\n\u0027test_delete_flavor_service_profile\u0027 no longer has to be skipped as\nLaunchpad bug 1843290 is now fixed.\n\nChange-Id: Ibe3ddbb18c289662940f442d6d75236f04a6b2cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/595e61d17cb27900546bdfc022121cb816283f6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/595e61d17cb27900546bdfc022121cb816283f6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
