)]}'
{"id":"openstack%2Fpatrole~690862","triplet_id":"openstack%2Fpatrole~master~I402a7ab2711e90d9c8f2173a624e652d31d70ce4","project":"openstack/patrole","branch":"master","topic":"ussuri","hashtags":[],"change_id":"I402a7ab2711e90d9c8f2173a624e652d31d70ce4","subject":"Switch to Ussuri jobs","status":"ABANDONED","created":"2019-10-24 06:50:43.000000000","updated":"2023-04-27 19:18:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f988fbd880a1ec32f83d4aaeec1a6eb2afbed18e","_number":690862,"virtual_id_number":690862,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-24 08:46:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2019-10-25 00:53:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":-1,"date":"2020-03-15 23:54:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 08:46:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-25 00:53:05.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-03-15 23:54:23.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"73b5a1d4d4211bf28f05ec11576477ed05c1035f","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-24 06:50:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aeb92fa19826642bb144c9601532bf8c145f94d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 08:46:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/221915904fca442db571e62a2952618f : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/799208ff7ddb4c3a9ac88afbff721fc7 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efd0625e9fd940878e7cb69b09932aac : SUCCESS in 8m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8a5dee704d74facb2265e2bff15b16c : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7140f49ddf41d78512742aa5ad92e7 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b84fcb6bf84e410f9527c2a728a2c4fe : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26454c7bde04c28b86c0fbe93d0f24c : SUCCESS in 13m 21s\n- patrole-admin https://zuul.opendev.org/t/openstack/build/a6a32cacad74452681056f2e2787f9bb : SUCCESS in 1h 14m 47s\n- patrole-member https://zuul.opendev.org/t/openstack/build/ac223ffaed9e4be69950fb0a691d01f6 : SUCCESS in 1h 06m 48s\n- patrole-reader https://zuul.opendev.org/t/openstack/build/35645b2d3d114ed5a778b6184c2fbc25 : SUCCESS in 1h 19m 21s (non-voting)\n- patrole-member-stein https://zuul.opendev.org/t/openstack/build/32c421e749af4824889a3dd3a67d82fb : SUCCESS in 1h 18m 59s\n- patrole-member-rocky https://zuul.opendev.org/t/openstack/build/627f09a7d41943eaae804fa7ac992041 : SUCCESS in 1h 16m 18s\n- patrole-member-queens https://zuul.opendev.org/t/openstack/build/893df1b92f424c45b01e7eecd7a3aafa : SUCCESS in 1h 27m 44s\n- patrole-multinode-admin https://zuul.opendev.org/t/openstack/build/37db2f576f6b49d0894f2bc2f4d6996b : SUCCESS in 1h 03m 26s (non-voting)\n- patrole-multinode-member https://zuul.opendev.org/t/openstack/build/626cbbc784c14e64860d3c2cc4fbb570 : SUCCESS in 1h 06m 20s (non-voting)\n- patrole-extension-admin https://zuul.opendev.org/t/openstack/build/461f0b16852b4109b5c33819c1c9bf69 : FAILURE in 37m 29s (non-voting)\n- patrole-extension-member https://zuul.opendev.org/t/openstack/build/ebe85b0b0eb547ba9cb7e77e9eff6bdb : FAILURE in 43m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b14e3933d37fe2cdad2dc1869d264de217c97f3","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-10-25 00:53:05.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"ab5eefe43abb5e4bf4ed785c76991cbecce24a0e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-15 23:54:23.000000000","message":"Patch Set 1: Code-Review-1\n\nthis is already covered in https://review.opendev.org/#/c/705525/","accounts_in_message":[],"_revision_number":1},{"id":"f988fbd880a1ec32f83d4aaeec1a6eb2afbed18e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-27 19:18:55.000000000","message":"Abandoned\n\nThis project is retired, for more information, refer to the https://github.com/openstack-archive/patrole","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"946656f32043fefc91bfde6960ee3f34e7859ec4","revisions":{"946656f32043fefc91bfde6960ee3f34e7859ec4":{"kind":"REWORK","_number":1,"created":"2019-10-24 06:50:43.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/62/690862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/patrole","ref":"refs/changes/62/690862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/patrole refs/changes/62/690862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/patrole refs/changes/62/690862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/patrole refs/changes/62/690862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/patrole refs/changes/62/690862/1"}}},"commit":{"parents":[{"commit":"45f01b741754f91023f7038851fc018c8fbd538d","subject":"Merge \"Fix missing \"raise\" before cls.skipException\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/45f01b741754f91023f7038851fc018c8fbd538d"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-24 06:50:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-24 06:50:40.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: I402a7ab2711e90d9c8f2173a624e652d31d70ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/946656f32043fefc91bfde6960ee3f34e7859ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/patrole/commit/946656f32043fefc91bfde6960ee3f34e7859ec4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
